)]}'
{"id":"openstack%2Fneutron-lib~931549","triplet_id":"openstack%2Fneutron-lib~master~Ide7933ca6debfebbf2ccc51a5dbdba1be638eeec","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-10 15:11:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ide7933ca6debfebbf2ccc51a5dbdba1be638eeec","subject":"Add constant for neutron-ovn-agent process name","status":"MERGED","created":"2024-10-05 13:01:51.000000000","updated":"2024-10-10 15:12:45.000000000","submitted":"2024-10-10 15:11:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931549","meta_rev_id":"2a44e4bc5f3d19c137a60bb174e24719dd9d80f8","_number":931549,"virtual_id_number":931549,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-10 15:11:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2024-10-10 13:03:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-10-06 02:26:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2024-10-10 13:03:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-05 15:28:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-06 02:26:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-10-10 13:03:08.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"65aa484eece65200e36845bc269fa1ae06417c52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-05 13:01:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48f33bd4944b1238ab038950b467d7132a0c6643","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 15:28:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c94c6a3257f44af8fe689b7403158a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efc5d33efd9046b0994810b3de68cbff : SUCCESS in 9m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fae54ae6d65240d38e4be6ea24f80a7a : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/326f6f1bbbc04bf6b7b3920ec9f72791 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4824f5abe9344e2b9e82bb3071b4762 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5dbbd5cdb0f4f87a9e14f396ef9a0e5 : SUCCESS in 6m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2056de27268446dfa6d92c7748a89237 : SUCCESS in 1h 02m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dd26aafa9db14f77af9445c5620dc118 : SUCCESS in 1h 27m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/873505383c804c30bb244ed46a410c30 : FAILURE in 2h 24m 26s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c7e2d629349347dba39d77646ed3e077 : SUCCESS in 1h 44m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/35f8ed50d9894178aab103c9c86e70a1 : SUCCESS in 1h 43m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b3ad4380adb14ae5a779e8502f30a185 : SUCCESS in 35m 05s (non-voting)\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/b9b2b0196a9747409fbebbe5e47497e0 : SUCCESS in 22m 29s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/39bd661497834d1b934cb12d2558d986 : SUCCESS in 5m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5fd392a9063d465a997c87632a69699b : FAILURE in 53m 16s","accounts_in_message":[],"_revision_number":1},{"id":"097741ec5081e176509f748728035349d6eb31fb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-10-06 02:26:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3d6bc691d41c71b10c4f1c9dd7c658bcdf88102","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 04:45:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f06c2389a6ea4b78901e4229775507bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7065abdd52d448d5899c409ca9df4d7f : SUCCESS in 7m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57e96d18a1154a8db05593a17c8ece7a : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ef526471694431a92e3cbf5385dab39 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b8af5fe7b8e4ef191151489c6e423ea : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fbae5ccb156746cdaff71ca3d7fb2b67 : SUCCESS in 5m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d93dec95d3154307a526e182bcba14d4 : SUCCESS in 1h 34m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cafbe4f52d8a49358046e86ac5d1f6c0 : SUCCESS in 2h 02m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a215cd1928a448998483631b2d617717 : SUCCESS in 1h 41m 02s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/465c8fb947624a339e413e91d4c9c1d3 : SUCCESS in 1h 54m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/98dfd0973a1e49bab148e9982be5f51a : SUCCESS in 1h 53m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9c32995a5c2246fa9cac7f0e3dbc01af : SUCCESS in 29m 39s (non-voting)\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/57cc9782cba54a88ac4fd5bb32be3754 : SUCCESS in 19m 58s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e1c5691c8ba2474a84a279db55a201b0 : SUCCESS in 5m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3cc432335594e8dbf7731c919bdb833 : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d4b53969c3d675a58862e3ae5b6e5a939c92a6cb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-06 15:11:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c2be58cbe033bc3a958a63412b47f90e3b09c63","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-06 15:16:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c1b89d8d468e442ed11f1ce27320ac00b3d01ff0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-10 13:03:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"31a9e3b2e3a6b05872dac3ef0b884db843f37762","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 13:04:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57a96713077f5946a24d96f253619cedb4a40bf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 15:11:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/222442e03c88423b84621d6a2077bf76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a4657f728fb4559b68e6bcd14feff2f : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e941c6da174476db6faf2ca53e42742 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd7f856cf365496faf06d343c7f99736 : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5495c92c390432ab357539f66b3c1d1 : SUCCESS in 3m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff152a041a5d42a3ad5fd09aebc73555 : SUCCESS in 1h 43m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5880bcb0e0a941aabeb255afb3c8c090 : SUCCESS in 1h 33m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2d2831658e7a40178b1a81facb933339 : SUCCESS in 1h 42m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c3bd2095cafe44bda5f41db931caaffc : SUCCESS in 2h 00m 20s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c280e80d51b841fa97605b749264a028 : SUCCESS in 1h 59m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/84f6ed15bbb641e5b0ff4434d968dda9 : SUCCESS in 1h 09m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7a7325566999857dd9e0a9f97800d8cbea320d28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 15:11:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2a44e4bc5f3d19c137a60bb174e24719dd9d80f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 15:12:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f964d2592b654219b2a81cdf58843e52\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53604336753a4d2092e287da23ae2286 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"acfda09015fa2f65c246ed6a0f78bb3bb925ba45","revisions":{"acfda09015fa2f65c246ed6a0f78bb3bb925ba45":{"kind":"REWORK","_number":1,"created":"2024-10-05 13:01:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/931549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/49/931549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/931549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/931549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/49/931549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/49/931549/1"}}},"commit":{"parents":[{"commit":"2bc06ffe7913c86a15ddbd404f94f08d8ac41845","subject":"Merge \"Deprecate get_notifier(service\u003dNone)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2bc06ffe7913c86a15ddbd404f94f08d8ac41845"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-05 13:00:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-05 13:01:37.000000000","tz":540},"subject":"Add constant for neutron-ovn-agent process name","message":"Add constant for neutron-ovn-agent process name\n\n... to migrate the definition still left in neutron.\n\nChange-Id: Ide7933ca6debfebbf2ccc51a5dbdba1be638eeec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/acfda09015fa2f65c246ed6a0f78bb3bb925ba45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/acfda09015fa2f65c246ed6a0f78bb3bb925ba45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
