)]}'
{"id":"openstack%2Fneutron-lib~931568","triplet_id":"openstack%2Fneutron-lib~master~If351db13fcf4a7401166ed64673ea4fdb4f7d778","project":"openstack/neutron-lib","branch":"master","topic":"abc","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-10-08 15:55:13.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}}},"hashtags":[],"change_id":"If351db13fcf4a7401166ed64673ea4fdb4f7d778","subject":"WIP refactor: use abc.ABC as base class","status":"ABANDONED","created":"2024-10-05 15:43:00.000000000","updated":"2024-10-09 21:25:31.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4e29195d2b0ebc89cd672b3dc65ddbf8e426c50c","_number":931568,"virtual_id_number":931568,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-05 17:36:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":-1,"date":"2024-10-06 02:25:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":-1,"date":"2024-10-08 15:55:13.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-05 17:36:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-06 02:25:51.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"5126e655cdd12983ee2950bc9e376c730b3756ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-10-05 15:43:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a10639f0fc7af3d7c5a5c624cadd4f6fde0b62f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 17:36:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ff248eb69394f419ce0bd9bc45f70e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94cc3274f3ec49b59b92860ac5cbf01f : SUCCESS in 6m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a93b1823e3443408ea6a280171aa975 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f2cd9fbe6164b6cadcef0ab2846c1b7 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/297619e31ad045b8872c920bb16dd9c1 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a167db183a2640338908e129039bc898 : SUCCESS in 3m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba31e6719b9f40e5bffce4e98fa49612 : SUCCESS in 1h 40m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2d9561a6f5f0471f90138ffc7614f2b7 : SUCCESS in 1h 33m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/785eada7d3a04cc490748e4f4d160d1f : SUCCESS in 1h 51m 38s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/7349f37f7b9645c382d906507b6aacd7 : SUCCESS in 1h 51m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ecab95bfdc5a4ccbb4b94068653db60b : SUCCESS in 1h 38m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb32749579364365b2f4ffa1963ba48c : SUCCESS in 30m 11s (non-voting)\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/3f4d68109499492ea2b25f2a3cb20846 : SUCCESS in 19m 27s (non-voting)\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ecc45a873adf4499957846e472f3f83b : SUCCESS in 3m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4f4bb904ff4944e9b0a26184fbe8b45b : SUCCESS in 57m 58s","accounts_in_message":[],"_revision_number":1},{"id":"fa3041d0fed933d8d9060debce2f5d9f7bc24d44","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-10-06 02:25:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"839226e83e2dafce34b533ec6a5c88e5ba813877","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-10-08 15:55:13.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4e29195d2b0ebc89cd672b3dc65ddbf8e426c50c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-10-09 21:25:31.000000000","message":"Abandoned\n\nI disagree with -1ing for using abc.ABC, but I don\u0027t want us to waste time on this further. I will revert to metaclass\u003dabc.ABCMeta in my other patches.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03062b2f1d44ad4b19407df924f53fd313c92af2","revisions":{"03062b2f1d44ad4b19407df924f53fd313c92af2":{"kind":"REWORK","_number":1,"created":"2024-10-05 15:43:00.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/68/931568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/68/931568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/931568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/931568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/931568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/68/931568/1"}}},"commit":{"parents":[{"commit":"be03db0bf143e1d1e8706ac0246acbaf24229a34","subject":"Merge \"Add new API extension ``uplink-status-propagation-updatable``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/be03db0bf143e1d1e8706ac0246acbaf24229a34"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-10-05 15:38:36.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-10-05 15:38:36.000000000","tz":-240},"subject":"WIP refactor: use abc.ABC as base class","message":"WIP refactor: use abc.ABC as base class\n\nTODO: enforce with hacking.\nChange-Id: If351db13fcf4a7401166ed64673ea4fdb4f7d778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/03062b2f1d44ad4b19407df924f53fd313c92af2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/03062b2f1d44ad4b19407df924f53fd313c92af2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
