)]}'
{"id":"openstack%2Fneutron-lib~931680","triplet_id":"openstack%2Fneutron-lib~master~Ife1abbf48c15a2440c8b1f5316c541781964ce5c","project":"openstack/neutron-lib","branch":"master","topic":"2025.1-neutron-jobs","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-10-07 19:51:14.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-10-08 19:16:33.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ife1abbf48c15a2440c8b1f5316c541781964ce5c","subject":"Use py312 for all neutron-lib jobs","status":"MERGED","created":"2024-10-07 16:04:02.000000000","updated":"2024-10-09 00:09:15.000000000","submitted":"2024-10-09 00:07:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931680-2025.1-neutron-jobs","meta_rev_id":"890dd52ada9d825c0952b7bd96d1c384844abbd9","_number":931680,"virtual_id_number":931680,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-09 00:06:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-10-08 14:14:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-10-08 15:06:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-10-07 19:51:14.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-10-08 15:06:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":1,"date":"2024-10-08 19:16:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-07 16:55:19.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"CC"},{"updated":"2024-10-07 19:51:14.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-10-07 20:26:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-08 14:14:53.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-10-08 15:06:51.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"61ad5b5ef58914a1ca232659a95daad578f73b61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-10-07 16:04:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc9a14f7f169ccc387f358cfb0d26ae63d88a06a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-10-07 16:38:40.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"6e90f16fd61a3b5c0b303aaf437476d4f9f95a5d","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-10-07 16:55:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4f54602530c3864bf0702ab33e44caa34c1eaef","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-10-07 18:28:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b05f2798f90251513f5f3f320590f144f7fa20ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-10-07 18:28:38.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"08b51c57e9041755bbee981bc3a43b46c3596c98","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-10-07 19:51:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"99f5a62f427e60b8f16d7add27c6b6bf1086f47f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 20:26:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a237cbcf3a34d77a9a5424b5a29ae43\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3f83f9d5b8554724ac4cfd4a11b2311e : SUCCESS in 1m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9098a746186c4d78abd0356ef7261cd7 : SUCCESS in 7m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c84fae818d1b402d953b9aa9e59f92f6 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/365351caa1ef48de8aa7ba737d5e54b9 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8cf14bbbc554a6998ff48018b1d05c5 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4f23041fdfb84cbb923cac815795590e : SUCCESS in 4m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66aaca2adceb4d8da0e79ae0a7ca2a7d : SUCCESS in 1h 13m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0fdb9cefd59b4cba8e37733f20648fcf : SUCCESS in 1h 55m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0a4b7c6d094e43ee9833df2f9fcc64bd : SUCCESS in 1h 29m 49s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/2393f56388d44a6baae311b3ca71dd6e : SUCCESS in 1h 45m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/10839d172afb463f96c9b7f6bc30fc67 : SUCCESS in 1h 28m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ae7aaa450ed4586b2d2e1d362e1adb6 : SUCCESS in 29m 51s (non-voting)\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/514fbef295c94ac6815233f7e3deb8df : SUCCESS in 20m 10s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e85e6653c617446ab38ad99ff37ef38f : SUCCESS in 4m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b795d77753e4d35bb8f850d830a0135 : SUCCESS in 45m 47s","accounts_in_message":[],"_revision_number":2},{"id":"8ef0f1070d4a5de4f9979c91e5805edc10112cf7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-08 14:14:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3b7f0cb0b6b7abf702265ad1a37a2f336a79abfa","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-10-08 15:06:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5f0262b501bcf2be95c5756a81cb60815b6eb8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 15:07:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"77e3be1f5372585bbc5a3018f242a2fbb47892e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 17:15:06.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d988e1c85ceb4851938ebd22ef8e1fd3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/26c51268b0e5483cb724427762189cb0 : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9b66584f0bb43e0a96eea9e03576623 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe43c259523c4e5d8461c0b3896fe5be : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e140eaa32fe4801accee31cfb97ba71 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8d8dd7fccb24aad9cffe8364493b558 : SUCCESS in 4m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4cd4360909d4536a3caf2cf81ee7d51 : SUCCESS in 1h 42m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/951ebd23840047088fa9eb35a642d0cf : SUCCESS in 1h 24m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e22aee0f37464a22a54bfb314b6df3d8 : SUCCESS in 1h 31m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dd70641796874ed0af65d7c9ae50ec70 : SUCCESS in 1h 38m 11s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/fa65b1e50b3c4dcebdf9769987acbb16 : SUCCESS in 2h 07m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c62096c9072d423e92c4a5e822d298fb : FAILURE in 48m 30s","accounts_in_message":[],"_revision_number":2},{"id":"71dcbfed9502f3ad228f2222b865ceef37a684c0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-10-08 19:16:33.000000000","message":"Patch Set 2: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bffdbe7a4b7e116c02f11138bf15df7b92dc4b5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 21:36:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd0e23a833f340198c1d048684a08481\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e49ec6882cfc4a8eb80549bf3ea057d5 : SUCCESS in 1m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ae07e3feda47c09ccac3bf41467ec0 : SUCCESS in 4m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94ed9cd187d6404c8116c2a6f2c3b7b6 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/446f8de65a674ad1b01dbb91c2381492 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e981ff9c96b476fa20666f90d19b45f : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1bff99d138234efe98e797a34760a3e6 : SUCCESS in 4m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/979fe989805241c490009f1e33aa8a8f : SUCCESS in 1h 37m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b55b71bbf35242b9b24514466add9403 : SUCCESS in 2h 02m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eee1c5c1dfc441b49e31bf3a3ab6dc7b : SUCCESS in 2h 16m 44s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/777b413517cc49258ac899d0d763ef8d : SUCCESS in 2h 18m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/04fbf6c4bac346afacff4e1cb06fc630 : SUCCESS in 1h 26m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c3ec22e20cf5413881c90afd74253fb6 : SUCCESS in 27m 22s (non-voting)\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/7fd5aad63c69471fa24622f1cae6f72b : SUCCESS in 20m 10s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/802f80e5d13343af950fbdd87dd30fa5 : SUCCESS in 4m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/85273afcacf34a46bacccd64a6861d69 : SUCCESS in 43m 32s","accounts_in_message":[],"_revision_number":2},{"id":"a064fd9eb091abc371c90d9453c86b9121f36a68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 21:36:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f000b87732975f617ab517b227bcc93eccce576e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 00:06:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4baea7fe57045189967487f60527074\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/76bef18789c44400afde1605a9021275 : SUCCESS in 1m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4901162412c453e94d7dc0511a31489 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66bc4d7601f344588c930fe532ac4041 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e486d595fac74bc990e2b5ada0bbfa4a : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d1e6e36ff0fb4a189291178e54959d34 : SUCCESS in 5m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e59bbf093fba474082fff28b6bcb8f3a : SUCCESS in 1h 31m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e7631518ac24425b9fdce1ba38bb2044 : SUCCESS in 1h 23m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1460d70786b442e987ba861df1d480d9 : SUCCESS in 1h 42m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9e163d13e810494bbc2813079be593d1 : SUCCESS in 1h 32m 10s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/1056134e59ee45a08cde7c7f0942be4f : SUCCESS in 2h 28m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b1e09d3fc1624ae39baa71bcc452535a : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":2},{"id":"e6d52ec49f0e277b4c3be24f621615c006bbb09c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 00:07:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"890dd52ada9d825c0952b7bd96d1c384844abbd9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 00:09:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c3bb15b22e348a396dbbca9b17cb88f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17d4e8d75eda489ab08113fd407aa38e : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3a03b626426370e4f97f52386b503b0e2dfd7f47","revisions":{"bdf087685d22de598bfea9d9bb0796d0db0d737f":{"kind":"REWORK","_number":1,"created":"2024-10-07 16:04:02.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/80/931680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/80/931680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/931680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/931680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/931680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/80/931680/1"}}},"commit":{"parents":[{"commit":"be03db0bf143e1d1e8706ac0246acbaf24229a34","subject":"Merge \"Add new API extension ``uplink-status-propagation-updatable``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/be03db0bf143e1d1e8706ac0246acbaf24229a34"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-10-07 16:01:54.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-10-07 16:01:54.000000000","tz":-240},"subject":"Use py312 for all neutron-lib jobs","message":"Use py312 for all neutron-lib jobs\n\nThere is a job definition here for testing\nneutron-lib with the master branch of neutron.\nAs neutron is no longer testing py311 in its master\ngate for 2025.1, start using the py312 version\nof this jobs so we can move forward.\n\nAlso moved the sqlalchemy job to start using py312\nas well.\n\nAlso updated the testing runtime to include 3.12\nand removed 3.8 based on 2025.1 governance [0].\n\n[0] https://review.opendev.org/c/openstack/governance/+/926150\n\nChange-Id: Ife1abbf48c15a2440c8b1f5316c541781964ce5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdf087685d22de598bfea9d9bb0796d0db0d737f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdf087685d22de598bfea9d9bb0796d0db0d737f"}]},"branch":"refs/heads/master"},"3a03b626426370e4f97f52386b503b0e2dfd7f47":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-10-07 18:28:38.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/80/931680/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/80/931680/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/931680/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/931680/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/931680/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/80/931680/2"}}},"commit":{"parents":[{"commit":"be03db0bf143e1d1e8706ac0246acbaf24229a34","subject":"Merge \"Add new API extension ``uplink-status-propagation-updatable``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/be03db0bf143e1d1e8706ac0246acbaf24229a34"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-10-07 16:01:54.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-10-07 18:21:29.000000000","tz":-240},"subject":"Use py312 for all neutron-lib jobs","message":"Use py312 for all neutron-lib jobs\n\nThere is a job definition here for testing\nneutron-lib with the master branch of neutron.\nAs neutron is no longer testing py311 in its master\ngate for 2025.1, start using the py312 version\nof this jobs so we can move forward. Changed it\nto voting as it has been non-voting since\nJune 2024 and checking the past few months\nchanges have not shown any failures.\n\nAlso moved the sqlalchemy job to start using py312\nas well.\n\nAlso updated the testing runtime to include 3.12\nand removed 3.8 based on 2025.1 governance [0].\n\n[0] https://review.opendev.org/c/openstack/governance/+/926150\n\nChange-Id: Ife1abbf48c15a2440c8b1f5316c541781964ce5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3a03b626426370e4f97f52386b503b0e2dfd7f47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3a03b626426370e4f97f52386b503b0e2dfd7f47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
