)]}'
{"id":"openstack%2Fneutron-lib~934173","triplet_id":"openstack%2Fneutron-lib~master~I3f6ec176d2b0eae4586a65c7c64740e09d3d31d1","project":"openstack/neutron-lib","branch":"master","topic":"psql-your-time-is-up","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-11-06 20:50:18.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-11-06 20:50:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3f6ec176d2b0eae4586a65c7c64740e09d3d31d1","subject":"nit: Drop mentions of postgresql","status":"MERGED","created":"2024-11-05 18:42:56.000000000","updated":"2024-11-06 20:51:25.000000000","submitted":"2024-11-06 20:50:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"934173-psql-your-time-is-up","meta_rev_id":"10012cab89b8fbad4091667eeacbf4008ce41430","_number":934173,"virtual_id_number":934173,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-06 20:50:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2024-11-06 18:42:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2024-11-05 23:22:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2024-11-06 18:42:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-05 20:51:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-05 23:22:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-11-06 18:42:51.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"da9fbd9ea21a8db2dbab6e005158a8304d4c0d5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-11-05 18:42:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de9c7b0a4199d97aa490169275489451429103a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 20:51:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/635ce048cd104d0a925d55f67dfc39a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac1d86cab8594902ad5101c9c047fc11 : SUCCESS in 6m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68ed7a646c114031adc01eba1eeefcb6 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c446d4aaae54503b06240fd14fb5353 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d9586140bb94a6baa57077b4ebd5790 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4c0349532054da18d563ee481698fdb : SUCCESS in 3m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2e27617e4544a4bbe4a9de882676a53 : SUCCESS in 1h 44m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/62c6ccdffc034ee78efc68b5b5303ff4 : SUCCESS in 1h 45m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a12d8e8b334e423694226d6c91e2b84f : SUCCESS in 2h 07m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/091d0ac6d326476580fa6aba0ca7ec9a : SUCCESS in 2h 04m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/329f247edc234dfaa7ea645c53fe2ce8 : SUCCESS in 31m 53s (non-voting)\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/47f1d5cb2b204812930bb9b55f41f44c : SUCCESS in 20m 57s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e6ab8d7eb44c46ad970eb88511f75f8e : SUCCESS in 5m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/60ee8b43b744463b9294e2b735bb0621 : SUCCESS in 53m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3785799a6753b926894700d24e857998d250431e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-11-05 23:22:25.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6afe95c0fec1ac294b0acfce4ffd6a02d6cbbf4","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-11-05 23:28:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c5468416c92ee54fe5b6c0321992a3dfd5dcd22","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-11-06 18:42:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15db4809ae5674f135437e93c49966d4c2427965","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 18:43:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a21ef5aab4b04a78314daa3bdf69828370decf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 20:50:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a25884c3a5cd4dd2a6a0f59b38139e10\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eaa56fc1efa4fc99797ad55b2612d1f : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d5e01e860e1481cbaa3ff911c9c5520 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90864618359e44dfb169ef7f37dca9db : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d82fd425996841f1bffa10065e334bee : SUCCESS in 2m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de8fd1daf0a44261b2f59646de9b0381 : SUCCESS in 1h 56m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5117931888f344febc62ab789d88d26b : SUCCESS in 1h 18m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a81a3ced846343f8bc7a440466b2d789 : SUCCESS in 1h 50m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2fff59dcb9734f86bb116d014a52a1af : SUCCESS in 1h 35m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fdf4146584414f48a6a06ad1660006a4 : SUCCESS in 56m 24s","accounts_in_message":[],"_revision_number":1},{"id":"1da2e1c44836d8689d6a77df4610df18a8789c96","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 20:50:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"10012cab89b8fbad4091667eeacbf4008ce41430","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 20:51:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4cb8cb332774aa486d47a4d1218d885\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6993b0db15504a2bac147894640b76bd : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fee4b2389fe033eb893f5e68371b4ae4312efc1a","revisions":{"fee4b2389fe033eb893f5e68371b4ae4312efc1a":{"kind":"REWORK","_number":1,"created":"2024-11-05 18:42:56.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/73/934173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/73/934173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/73/934173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/73/934173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/73/934173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/73/934173/1"}}},"commit":{"parents":[{"commit":"b60e363e5ec00f24a7c648110a2d35539351a9f0","subject":"Merge \"Add note about requirements lower bounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b60e363e5ec00f24a7c648110a2d35539351a9f0"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-11-05 18:42:26.000000000","tz":-300},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-11-05 18:42:26.000000000","tz":-300},"subject":"nit: Drop mentions of postgresql","message":"nit: Drop mentions of postgresql\n\nNeutron is dropping code related to the backend.\n\nTrivialFix\n\nChange-Id: I3f6ec176d2b0eae4586a65c7c64740e09d3d31d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fee4b2389fe033eb893f5e68371b4ae4312efc1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fee4b2389fe033eb893f5e68371b4ae4312efc1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
