)]}'
{"id":"openstack%2Fneutron-lib~934587","triplet_id":"openstack%2Fneutron-lib~master~Ia480dbb113d407312f9a82794826da9ba5fce463","project":"openstack/neutron-lib","branch":"master","topic":"policy-yaml-file","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-11-20 12:30:21.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-11-20 12:30:21.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-11-19 17:23:50.000000000","reason":"removed on reply"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-11-20 12:30:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia480dbb113d407312f9a82794826da9ba5fce463","subject":"Remove default override for config options policy_file","status":"MERGED","created":"2024-11-11 05:50:44.000000000","updated":"2024-11-20 12:31:29.000000000","submitted":"2024-11-20 12:30:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934587-policy-yaml-file","meta_rev_id":"19c4374327c3bfac48d67209dddc722d8f22571f","_number":934587,"virtual_id_number":934587,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-20 12:30:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2024-11-20 10:12:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-11-19 17:23:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2024-11-20 10:12:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-11 06:23:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-12 21:54:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-11-14 09:16:52.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-11-20 10:12:17.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"35d0348114be3ffbff05a967a845c5b7a6a1cd52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 05:50:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98d392705cd59dd55ce283b6e314204a6aae16ab","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 05:51:31.000000000","message":"Topic policy-yaml-file removed","accounts_in_message":[],"_revision_number":1},{"id":"51110c7608cf9422d3592e7671188939fa464c84","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 05:51:46.000000000","message":"Topic set to policy-yaml-file","accounts_in_message":[],"_revision_number":1},{"id":"53012d0fcf3e6304be37b6d70aa80e9635e6cd11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 06:23:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8b0d87383e24a69a1dfa3e1583a7c7c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/663615f3579f4aaebfc71872f8998a41 : SUCCESS in 1m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24024589ce8447438e22a253fdf4e51a : FAILURE in 3m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82ee732a741c49f9a4f82426da4a3219 : FAILURE in 1m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66a0834aec964fc5ab3a09b2d2bbda9b : FAILURE in 2m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6543dcd61ee44d3e934024d5f91f6812 : FAILURE in 2m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ba8831076414b278953fc8436481f2f : FAILURE in 2m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1518725310c14d429f27ffd3d1d23945 : FAILURE in 20m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7063f139b79a47edba95c5014f7eda2d : FAILURE in 8m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/64c68b8c118e44279f76b4142d4a2a1a : FAILURE in 8m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9b769205416b406c8345436c0073bd1a : FAILURE in 8m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ed78a255065f4f6b826e44c404dcb17e : FAILURE in 11m 16s (non-voting)\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/dcd44036b29e4b09859354c7acdb2e0a : FAILURE in 4m 19s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/53acb5c0ca8a41fc8f5f9a68859a4d9f : FAILURE in 2m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2029c7e756284c5288bec8ad3bfaaa74 : FAILURE in 21m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2a63285bc825e2eed90822a7886703316da20ba9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 21:19:29.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ae3e28a9becd31b7efa36a66687cdef919784351","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 23:13:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/497c6ee26121402fb36becccca03357e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4c8e075800c848fd966b676b714013d5 : SUCCESS in 1m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/617016914f3b4168bdcd6b3d6cced3b7 : SUCCESS in 4m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb98269db1a44f23b0627665431cda7a : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6f673d4026449c088b4a0b5407c63e9 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af52a69916db49c6bc61d6d567a42877 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5610dec5850e4f2fb587ddbb71dce717 : SUCCESS in 4m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed0c1ef3fbe04a26a901c1173e1816c2 : SUCCESS in 1h 21m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b139632e6a99454eafaa267b80f67621 : SUCCESS in 1h 52m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6697545675824a3aa963ed8e7e22f4be : SUCCESS in 1h 33m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6317fc0a9bdd420da50f81b0de47efc1 : SUCCESS in 1h 39m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/132dbd1d8b524799b9e2c6380e89a686 : SUCCESS in 30m 03s (non-voting)\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/92cfc19b44e14e849d4aa952dbbcbf3a : SUCCESS in 21m 24s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/908def490e454529aa46fc4d6a05fbf8 : SUCCESS in 6m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b967c2a858d94127bede017e361fc0e4 : SUCCESS in 52m 24s","accounts_in_message":[],"_revision_number":2},{"id":"a7cd76f75d553ff286c42d0ce1664e7daf082702","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-11-12 21:54:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"395343df01aaa888dd63a29c5a87d1def4c3e563","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-11-14 09:16:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"30932b8d5b1bac5131924f0cde0718708e181d55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 09:17:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4a0e61e31e69d83e112b95a198a84abb29b974fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 12:27:09.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/12abb3bb9d124b80b6ad94ba7884d806\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/55f663854a6a48a08c909321de68fbd8 : SUCCESS in 1m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a7f390db824dd89eeb411d58809b8d : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6f08be4d49c4e6387a2abacdd40c2fd : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d70ba598b8d449618f4339210e2239ec : FAILURE in 5m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63164a78f9724966838b9b6d630f2f4b : SUCCESS in 2m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0846345e4d194abc9e75e5dc5a7bd3df : SUCCESS in 1h 36m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f821aba1b51b4141b264598583f6a1dc : SUCCESS in 1h 38m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a9d6cd2a07f4401992d868d8a353c667 : FAILURE in 3h 03m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/56d92777766648aea5111fcae51dbf6a : SUCCESS in 1h 53m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d14af312443446d4ab480b55330ee2cb : SUCCESS in 55m 26s","accounts_in_message":[],"_revision_number":2},{"id":"3f1019f0b9146b806e8d8d948db7d2601ad6d953","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-11-14 20:24:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"15167e8ca3d04c3f2afb931b81b574e9348eed5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 22:22:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/26ee2e10085d454a9118316bda7b6dff\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/83f2d778e514479bb90014e946f62dda : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c40e38727f6149248191988cd64c6c59 : SUCCESS in 8m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21cc8b6c674c402e8bc60c1a47e18c18 : FAILURE in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5df1b0524bf400a94dbb8deb5aeb28e : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa34b89888e7444588d42727a511a424 : FAILURE in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/709bec5231a440b1a9d1d997bdac56fe : FAILURE in 3m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84bd6eb2f30f49ef92cba3db26a815b1 : SUCCESS in 1h 07m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d336b63e75614b499b76895292b0470e : SUCCESS in 1h 40m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2b547b6b066f40329701fdda0eb06f62 : SUCCESS in 1h 55m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a6faf3df5ef7414abbcad1d489950608 : SUCCESS in 1h 56m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8c879280407b4d909a96fe8b6fb96590 : SUCCESS in 27m 24s (non-voting)\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/9407a25151bf418e88d9e4921c2b5f2a : SUCCESS in 24m 02s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0b74a95d5942450d82ed5a9e2d011f27 : SUCCESS in 6m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e84a02fe47b9491db080309eb252068b : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":2},{"id":"aab9cfd88cc69cfb1224a3a4f1d23b533cfc0313","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-19 11:51:47.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5061235d401723774d7ca9b04323e0a092e9a94e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 13:58:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efbfce9c33474626bc3a659895dac040\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f509033729234261ad6aff2eb77e9648 : SUCCESS in 1m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e6df1a24134dc282b7344908bd8835 : SUCCESS in 7m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee31b7ba0dbc4db195c4286ef4808e61 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc1e7a6fe14e4d18b544d947269690e4 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b53b192e31cf4aed89829d61d4998c86 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d1b396436e8c43c0b5447aaaa6a0e182 : SUCCESS in 4m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31c8336eda71436d8f4da6805771af0a : SUCCESS in 1h 31m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/aaee3455819541a4a29e0d361dffe8e4 : SUCCESS in 1h 38m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ee4fe4efc6134e629c113392ae2e21ce : SUCCESS in 2h 04m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b97d1721ae1147b0a57903685cc1973e : SUCCESS in 1h 53m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3d8c40773b944f52ab8ffb5d9f0110e1 : SUCCESS in 37m 35s (non-voting)\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/351ffd34f4d94d5ebb3c5314ab872d28 : SUCCESS in 19m 01s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ef51234ea5cb47dea4d9efee8478578d : SUCCESS in 6m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/381504ef5c7e4da3a2575ca0e9bebdb8 : SUCCESS in 53m 56s","accounts_in_message":[],"_revision_number":3},{"id":"1459fd94ba698dd24b929260ef14d5ed815389c4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-11-19 17:23:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1d636ab4593ff67aca2bf90b7f9215a4894e6350","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-11-20 10:12:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c44174ca4f104ae1dbaff2d0d1d1f21f4b51b162","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 10:12:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"76d16f6f37e5174d924d294e311eb53a4dae04e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 12:30:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8490353a178448a487ce0ec11f761832\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/28715ea95c4544c184d291ee082a8fa7 : SUCCESS in 1m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5469b26b471c45fcab314d3143436896 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cfd48ace56c4dbeb6cae576e224aba8 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1205cc05ed884236967fb27b5d418fd3 : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c275f7c42feb4aad92597ee39a842ae8 : SUCCESS in 4m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2100eb73db7547c5903f1dba086f858b : SUCCESS in 1h 38m 10s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/668d2d3d8a504dd99cbc09753c4cb247 : SUCCESS in 1h 43m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9631b28e19af42f1a42d165eaaac998d : SUCCESS in 2h 16m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eef835f686584253898fd5b996992313 : SUCCESS in 1h 40m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/26dc9671942b4544ae9ca85bdc61b7db : SUCCESS in 44m 23s","accounts_in_message":[],"_revision_number":3},{"id":"b36b363be4fb92a705c84c7de730bcd47b7881d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 12:30:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"19c4374327c3bfac48d67209dddc722d8f22571f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 12:31:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83314fee14234f3489f968281ab5862a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d53b8af8a87a42079f9dde01d30df505 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"86efeab80fb08ce50540df54ad1dc3195fee44ea","revisions":{"68beb4f038a1d9d9dbf8af17c60ba86821347248":{"kind":"REWORK","_number":1,"created":"2024-11-11 05:50:44.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/87/934587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/934587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/934587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/934587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/934587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/934587/1"}}},"commit":{"parents":[{"commit":"c4baf59bc1bac0ef5b037cb9d6bd8de6944bab0e","subject":"Merge \"pyupgrade changes for Python3.9+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c4baf59bc1bac0ef5b037cb9d6bd8de6944bab0e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:49:27.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:50:13.000000000","tz":-480},"subject":"Remove default override for config options policy_file","message":"Remove default override for config options policy_file\n\nolso.policy 4.5.0[1] changed the config options policy_file\ndefault value to \u0027policy.yaml\u0027, which means it is changed\nfor all the OpenStack services and they do not need to\noverride the default anymore.\n\nNOTE: There is no change in behaviour here, oslo.policy provides\nthe same configuration that services have overridden till now.\n\n[1]\n https://review.opendev.org/c/openstack/releases/+/934012\n https://review.opendev.org/c/openstack/requirements/+/934295\n\nChange-Id: Ia480dbb113d407312f9a82794826da9ba5fce463\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/68beb4f038a1d9d9dbf8af17c60ba86821347248"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/68beb4f038a1d9d9dbf8af17c60ba86821347248"}]},"branch":"refs/heads/master"},"6b47dc4c0ed1cd8860300105ffac2887554229ab":{"kind":"REWORK","_number":2,"created":"2024-11-11 21:19:29.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/87/934587/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/934587/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/934587/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/934587/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/934587/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/934587/2"}}},"commit":{"parents":[{"commit":"c4baf59bc1bac0ef5b037cb9d6bd8de6944bab0e","subject":"Merge \"pyupgrade changes for Python3.9+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c4baf59bc1bac0ef5b037cb9d6bd8de6944bab0e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:49:27.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-11-11 21:19:29.000000000","tz":0},"subject":"Remove default override for config options policy_file","message":"Remove default override for config options policy_file\n\nolso.policy 4.5.0[1] changed the config options policy_file\ndefault value to \u0027policy.yaml\u0027, which means it is changed\nfor all the OpenStack services and they do not need to\noverride the default anymore.\n\nNOTE: There is no change in behaviour here, oslo.policy provides\nthe same configuration that services have overridden till now.\n\n[1]\n https://review.opendev.org/c/openstack/releases/+/934012\n https://review.opendev.org/c/openstack/requirements/+/934295\n\nChange-Id: Ia480dbb113d407312f9a82794826da9ba5fce463\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6b47dc4c0ed1cd8860300105ffac2887554229ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6b47dc4c0ed1cd8860300105ffac2887554229ab"}]},"branch":"refs/heads/master"},"86efeab80fb08ce50540df54ad1dc3195fee44ea":{"kind":"REWORK","_number":3,"created":"2024-11-19 11:51:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/87/934587/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/934587/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/934587/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/934587/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/934587/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/934587/3"}}},"commit":{"parents":[{"commit":"b4745594a2970ddd749b3b49566f7e84586f3850","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b4745594a2970ddd749b3b49566f7e84586f3850"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:49:27.000000000","tz":-480},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-19 11:51:00.000000000","tz":0},"subject":"Remove default override for config options policy_file","message":"Remove default override for config options policy_file\n\nolso.policy 4.5.0[1] changed the config options policy_file\ndefault value to \u0027policy.yaml\u0027, which means it is changed\nfor all the OpenStack services and they do not need to\noverride the default anymore.\n\nNOTE: There is no change in behaviour here, oslo.policy provides\nthe same configuration that services have overridden till now.\n\n[1]\n https://review.opendev.org/c/openstack/releases/+/934012\n https://review.opendev.org/c/openstack/requirements/+/934295\n\nChange-Id: Ia480dbb113d407312f9a82794826da9ba5fce463\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/86efeab80fb08ce50540df54ad1dc3195fee44ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/86efeab80fb08ce50540df54ad1dc3195fee44ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
