)]}'
{"id":"openstack%2Fneutron-lib~936265","triplet_id":"openstack%2Fneutron-lib~master~I180d6fef6f65923931bf1d3c0f46626123ac7d7f","project":"openstack/neutron-lib","branch":"master","topic":"bug/2089673","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-11-29 14:21:53.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-12-02 07:27:22.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-11-29 14:24:35.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}},"hashtags":[],"change_id":"I180d6fef6f65923931bf1d3c0f46626123ac7d7f","subject":"Switch to neutron-functional job","status":"MERGED","created":"2024-11-26 14:27:18.000000000","updated":"2024-12-02 11:42:24.000000000","submitted":"2024-12-02 11:40:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936265-bug/2089673","meta_rev_id":"be531819f922041c055ee6a6891911927959f66c","_number":936265,"virtual_id_number":936265,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-02 11:40:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-11-29 14:21:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-11-29 14:24:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2024-11-27 12:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-11-29 14:21:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-11-29 14:24:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-26 16:23:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-26 21:09:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-11-27 12:00:28.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-11-28 05:53:30.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-11-28 05:53:30.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"93cfc86d22821f4cdef4e784fd9ce1eb37c43fbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-11-26 14:27:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08683232eda28ec176e5b9f3444909a8f571c938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 16:23:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db393003c9a54801a53e467891857b81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57286420c32847f888e882f48e7afee3 : SUCCESS in 6m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecdb0a51e4a44a85a50209b2785fa383 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/505319da57d24f6a8755381d24abf3e5 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d898dfa046654d1cacd53c3932f413ed : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/413f82ac93a54f6c965a6d53527fabcb : SUCCESS in 4m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bcb7d45d2a63488f9cc831713951325b : SUCCESS in 1h 49m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e853d8bc75584ba8a325d730a1440881 : SUCCESS in 1h 24m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/62c0c32f115947d58fb4d4c9059e5495 : SUCCESS in 1h 39m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f5b55a72fc654236b5d3f402762ea72a : SUCCESS in 1h 54m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c0f949f51f914427b92ab37741df781f : SUCCESS in 30m 19s (non-voting)\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/cd02b5790dd74e08adf11ed11a4a300e : SUCCESS in 22m 07s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/65668d7f86dd4a6caceab487259dc93e : SUCCESS in 6m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bf32586e4db94811b230ee09b009c0f9 : SUCCESS in 53m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8313e5dde313ad6c9c0c7889f0a1e019c0278837","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-11-27 12:00:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"28e10da5e39c48774ebbbc9e441b9ddb04a663a8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-29 14:21:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cbff5d47fcdd930df1319e02473496aba93a395c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 14:23:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ad3d285f105c4d779213e6c33c23350eef2911d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-11-29 14:24:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ad7b34d004607e818fdadf8beef04d30d9d21128","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 17:09:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0e8f46e6083415b901eb367d8371dc1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef2828cbd7d416781ce34d2182d8b3c : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63396098a6244254a487c8134a2b495f : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42898509fae64ba9a6281c1721d9d03b : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1e40cd7b7e93442aab801629775bddd6 : SUCCESS in 4m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c05ec108b14467fa6e131cdb2a1de05 : SUCCESS in 1h 27m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5308fe43c211475591377f30ce645d2c : FAILURE in 2h 44m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/49a6ab7643e040908b68c706b80e2a3e : SUCCESS in 2h 01m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/27eae6232a3d4e41802f1371439be7fa : SUCCESS in 2h 18m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/375115ac7a774823944fe0e10b19635a : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":1},{"id":"27ae4f8d6b1a7d69366625223318e4b09d54bcf8","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-11-29 17:13:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6269b432617515f82494ee30a890e2ec517c49d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 19:15:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02d50fd46079485b934cfcce127d07c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be5adebe9fb94690b312df06aae52e3a : SUCCESS in 6m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6f0f8cd62834bae9b1d45aa468ca6a3 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b10dcb415094c138f96e7898386e4d7 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32ddc38b04da4fc584be6c064379f8f4 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/96faf1379cee44d38a2de4531c986575 : SUCCESS in 5m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc9d20f0cb5149748834f56ecf23d896 : SUCCESS in 1h 07m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/964688d56f9d4836981cede6d68f38b9 : SUCCESS in 1h 27m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c546e41b5cfb4dea9c25c5e5afbb23b6 : SUCCESS in 1h 43m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ef2e4c2b241e4cb2a260e2bb852f4f94 : SUCCESS in 2h 00m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d8961c4f47e94d328b684274d9540ea0 : SUCCESS in 27m 59s (non-voting)\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/eea7c535f91d4cc48c82ef111d4ddaf3 : SUCCESS in 18m 36s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6a83162da2e64a34b37ab4a95d098df8 : SUCCESS in 3m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c15accc253b64a68a4016b53417dd96b : SUCCESS in 50m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ea5356629aaf71bff77b14b9cc623cbbeb675d0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 19:16:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34f73712c68ed7b01f109dcbbc9e9d0864628f8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 21:27:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/91ea990ac81e4f4783c5f03e8193b215\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e77dfded59a3430bb9c94c295faae3ce : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13c7ed7e0bbd432ab35f96c74e272ab2 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b069a9ce70d944bdbb90364333f8502e : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c55473ccd1324584aaf45e749b0ba6cf : SUCCESS in 3m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/771e785d8186433ba71d84823f1153c5 : SUCCESS in 1h 51m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4eeb495e534f43e0b9b3fa0402213910 : SUCCESS in 1h 14m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/90acaa4840244f66a0befef107a89a45 : FAILURE in 2h 10m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d996399788ff4e7ea3775c042a0118a5 : SUCCESS in 2h 04m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3456d2dc37eb4f45b06ddf14025e1f82 : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1b0ba172c18fb507944af666dcc01acf50b63a61","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-12-02 04:56:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44ec2fcaf4bd1e7cde47d8a6bf2a7d38ad4b1ef0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 07:15:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/525b79a3c5134e72b62ba6e28be63861\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78568da8a7c44a598084def75c3d71f8 : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf0aa65567fa4f3d89c7661ec8a81887 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6543ab05f8564439a8d9e398a8d484af : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/332c9197d32848ab914cbb57e2d390de : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ef193049532401795a88bd992e06672 : SUCCESS in 3m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fce18c5791f46479ea0c0e2c5380bd1 : SUCCESS in 1h 45m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/90a4b2c7068b4e7ab4030b1d809a39ae : SUCCESS in 2h 00m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6e29e588ddc9424caa4509b2b2c41f45 : SUCCESS in 2h 17m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ce7e204bfedb487d8cf53f20d4a275a8 : SUCCESS in 1h 54m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9b14c36c4b5f4e3a8158850ff347ad1f : SUCCESS in 39m 28s (non-voting)\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/febbccf184f14b9fa0167e80986f1dde : SUCCESS in 21m 37s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/86be657752da4a55994e7a09a942d2cb : SUCCESS in 6m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5399a6b575a143cd95eb2ddd05eec8b9 : RETRY_LIMIT in 8m 55s","accounts_in_message":[],"_revision_number":1},{"id":"599bd313070bb80927e4d0e3b679e1a832dda8da","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-12-02 07:27:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1cafb711ba0fa7950369a11cc4f12af15d33f847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 09:33:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27afa19b41014b7e93bea199847f3609\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70da27921c8d4a679f0cac092e48436c : SUCCESS in 5m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a48a881de1074caa935f3466e959b2b3 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fb225245dd947b2b25d8d7fc70b5949 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b35c60bedacd4657b864d070bac94fea : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a38fedca2814a789d4f6ba0c4c15857 : SUCCESS in 2m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a24dd1580837429f9bd583779f582a8e : SUCCESS in 1h 01m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9504f7f62e3141a1abded55d607404a2 : SUCCESS in 1h 30m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/72a5157b2f814f1386f55eb7f9a52b02 : SUCCESS in 1h 48m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/103ae8b56f034e1e872734d027f1eec5 : SUCCESS in 2h 03m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ba02172bd4ba421085b8dc42eb2a07d4 : SUCCESS in 36m 06s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8234305eab6541609eedad181405d524 : SUCCESS in 1h 32m 03s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07a14d3abfd045f6af6cc0457e0365e8 : SUCCESS in 1h 47m 41s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/75c5a1e07ee040ad808cb72e9223fb54 : SUCCESS in 16m 49s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d02e337f31004c7382444eb1ceccd620 : SUCCESS in 5m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6c6a986105ab410bb7f890cc1174575c : SUCCESS in 45m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d360a19e0b7b6491905797abb5a160cc1a4432e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 09:34:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c04dbd3fa11024d81379b4ffb3017e73deca1b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 11:40:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20c9384609cd404bbef8912baea55649\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/595eb4a7a7b848bd8ce89f3ddaede94b : SUCCESS in 19m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5814f1e80214b83800a10e5a8e2d91c : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7eda2aa1e0d94ca29aeb75ee342ada43 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4f7f6b1e3a6e4e65990f79062a6ed2d1 : SUCCESS in 8m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/388ee5a864ad4b8a81915a8964e6d933 : SUCCESS in 1h 43m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1d7f244ea0c44af79beab484c6c04591 : SUCCESS in 2h 05m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/920d5b76e1cc40fcbc26b68e2f3add34 : SUCCESS in 1h 44m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c7a8b559b73e4a0bb62986925dfdf697 : SUCCESS in 1h 59m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7b8f1cba493349d0ada5f4ac07855a36 : SUCCESS in 53m 06s","accounts_in_message":[],"_revision_number":1},{"id":"9728e41e8460ce1b995994c204529cf5a91ba364","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 11:40:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"be531819f922041c055ee6a6891911927959f66c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 11:42:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fa383fce38b4622837578d0a4e4ace1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51a1fd5e5cbe46de95895d5828af84ed : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2fedaba77b902663c5d0064dfb569e7c977cfed","revisions":{"a2fedaba77b902663c5d0064dfb569e7c977cfed":{"kind":"REWORK","_number":1,"created":"2024-11-26 14:27:18.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/65/936265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/65/936265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/936265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/936265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/936265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/65/936265/1"}}},"commit":{"parents":[{"commit":"86efeab80fb08ce50540df54ad1dc3195fee44ea","subject":"Remove default override for config options policy_file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/86efeab80fb08ce50540df54ad1dc3195fee44ea"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-11-26 14:22:29.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-11-26 14:22:31.000000000","tz":330},"subject":"Switch to neutron-functional job","message":"Switch to neutron-functional job\n\n[1] dropped the uwsgi job as default switched,\nso let\u0027s run neutron-functional job now. Also\nset zuul_work_dir as needed with [2].\n\n[1] https://review.opendev.org/c/openstack/neutron/+/935626\n[2] https://review.opendev.org/c/openstack/neutron/+/936262\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/936262\nRelated-Bug: #2089673\nChange-Id: I180d6fef6f65923931bf1d3c0f46626123ac7d7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a2fedaba77b902663c5d0064dfb569e7c977cfed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a2fedaba77b902663c5d0064dfb569e7c977cfed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
