)]}'
{"id":"openstack%2Fneutron-lib~936841","triplet_id":"openstack%2Fneutron-lib~master~If6436c048bf51d503c4f433a900b67e9945965ac","project":"openstack/neutron-lib","branch":"master","topic":"bug/1915151","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-01-21 11:44:46.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-12-06 21:29:39.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"If6436c048bf51d503c4f433a900b67e9945965ac","subject":"Add definition of the \u0027qinq\u0027 api extension","status":"MERGED","created":"2024-12-02 14:54:14.000000000","updated":"2025-01-21 14:32:04.000000000","submitted":"2024-12-06 23:20:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936841-bug/1915151","meta_rev_id":"246ba28363ec452d54bce836e0826dcff9f0b1f0","_number":936841,"virtual_id_number":936841,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-06 23:20:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-12-05 09:41:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-12-06 21:29:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-12-06 21:29:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-02 16:42:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-12-02 17:27:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-05 09:41:21.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-12-06 10:34:51.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-12-06 21:29:39.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-12-18 22:16:05.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"CC"}],"messages":[{"id":"bbfcb6dae512f5270e48d157a60ced8a9b43b5c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-12-02 14:54:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57b8966c51558f564c6f738a88341c0a5d49d88f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 17:27:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2907187713004b6ba2e8ff6e6026a8ac\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5e5bf9a00dda411684b7e4164ec2b442 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdf3e0705c774b69b0b6ba81ad570278 : SUCCESS in 6m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1de7b97aa2934b2ca9c163d313ac8ce9 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d590427a675245f7b137912b6d18f472 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2f35e1e48774ca1b624dfed11ff03fd : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74be646548ea4e74976b8ddbf66f7bc8 : SUCCESS in 4m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c55ebcefe99b49ad976b09ff6cfcd54f : FAILURE in 32m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/037b1225efaa417eb144d9609763ebb7 : SUCCESS in 3m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/55821d4b725f47fbb3f1c7e6f3bbcf25 : SUCCESS in 2h 00m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dfdeef8521ab4395bd244a719e97d3dd : SUCCESS in 1h 59m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a23c78c577c3429e96d059421f174aa7 : SUCCESS in 2h 30m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d237629976da49c5845b130e0e7d9ecf : SUCCESS in 38m 14s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/661f75d1b1404f15ae9f7c7deaf289fd : SUCCESS in 1h 53m 42s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64fd89453a1a4cf3a71f26e8337006e7 : SUCCESS in 1h 52m 37s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/2a7ae0009fff4ad399cea53f2d17b6d9 : SUCCESS in 20m 52s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5dafc7b6ed0043ad948c5bb794322df6 : SUCCESS in 5m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/37863311cc3645cfb8e8749590add0d3 : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":1},{"id":"91c0adbc1e79be3bca6f2b5346301fe92c2044dd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-12-04 16:05:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1b80f552ad202a520448816210ac8fe5b26f441","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 18:35:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e634ada03a741eba05c8e8dd97389be\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/707d36127e1f4de7877ad6d0fd3277d6 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13979bfe6eaa4bc7963d97a58333d9aa : SUCCESS in 7m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91dbb86647dd46aba99ee703dce20951 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66548babf5034592ac78f5ca03a8563c : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed1c95294b1340f3beb025670cdc1a55 : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7b329b00d3524b9dbb26e9fe285b8885 : SUCCESS in 3m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa49ce3af31f42d89ce2da6c4bc7e09c : SUCCESS in 1h 46m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35d9ba01a77f4d8eb745c853abe58d86 : SUCCESS in 4m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/58d9a9664d034a4f9dc5a1dcf58bafc4 : SUCCESS in 2h 18m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f803bbe11cbe4a1388944c810c4b2240 : SUCCESS in 2h 26m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ebae479846d648cb8664f761d6339e56 : SUCCESS in 2h 04m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ef52d1fe38c64cb692d74f4a31664df9 : SUCCESS in 41m 40s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6de07afeaa5b45589986772b5019f5de : SUCCESS in 2h 03m 18s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98949d88c52b4ddfa2b8c2013dbdad83 : SUCCESS in 2h 02m 18s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/725fb4c00e314a6f8af2cf4364a0fc32 : SUCCESS in 21m 39s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2cbfec45277f4d45a38c68f23bb42590 : SUCCESS in 6m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8cd88b9c4e754d62b15becc4502cddb2 : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":1},{"id":"52743e46a88c6130ba57914a60a741a5e43e4f22","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-12-04 22:00:12.000000000","message":"Patch Set 1:\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7508afcff6afe86289e045963218040908db6683","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-12-05 09:33:45.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9ebbe11cd8fde0a901e075d15b35bf9592688121","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-12-05 09:33:53.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"86cd08444d4215f8d7fc928deebfc1e41f89540a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-05 09:41:21.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"93353e4b90e14bcae706e3b9bbbc0aaf74a234e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 12:36:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2aadbdc5e5944733a3a701e327822406\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3b54be2ad0994b6595b4235c6567f83b : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fcaa0618d3f4d17af46cd396e0acb6f : SUCCESS in 6m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0c7bc1c74cc43a18053e0873e3ea324 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b5a7e3774994640b865b2c8f82ccf78 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2490082c90aa4af1a73f2d4be3b6452f : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/67ffaa7103e94e6eb87901c9b4f61f1f : SUCCESS in 4m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/abb21f2ce3254c7aa031a256c30e628b : SUCCESS in 1h 28m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fffb07493c145c1ad2a597cd30deb06 : SUCCESS in 3m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5dc6a8ce70814e88907ca1e08db7fefe : SUCCESS in 2h 06m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6c0b3001be6e48cf9285fced8c0d2ad4 : SUCCESS in 1h 54m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/92d10b5894ef4605b7c7e3cee77af39b : SUCCESS in 2h 59m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6e4fd3864f81498181dcf05d31fea74c : SUCCESS in 48m 43s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0bc561b3ab434b8eaac89f3577ffa656 : SUCCESS in 1h 58m 12s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46c7b670b14b4356adf51d625510c6b5 : SUCCESS in 2h 27m 02s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/33cd9e2ff06e408b90c29a0344c3bc90 : SUCCESS in 20m 04s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/50253373923e4f28b34e125c18b74a65 : SUCCESS in 6m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/77c9c1d68778495d95bc118f80b2c2b4 : RETRY_LIMIT in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"45a8f77682188a9268d4c6b03c2b62cc22b85851","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-12-06 10:34:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"55c9eea7df67f6560b9103e24eccb7ff08429e2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 13:02:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd82594ac3884f6da560400a863a1227\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/75b87078a2ad4bb69bafab7ec0cf30c7 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d934336f11f4f8ab26ad45a1a9c1c48 : SUCCESS in 6m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e97dcaf796d401d95c99cc87d8e6eb5 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c56ce29df3fc4dc1ae28e288893ea3db : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2127c23d14a42428e26ae6af8dd9c43 : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb552b00bd0345fcbf6620fad499e728 : SUCCESS in 3m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43ebc7450b084d228cad515f104ecfe9 : SUCCESS in 1h 25m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4450fca28cc44cdd90a8bb65fd70aeb3 : SUCCESS in 3m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/920438c3a098489d9be617e873713be8 : SUCCESS in 2h 16m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/744bf39c22a241a581e08401f70bdca7 : SUCCESS in 2h 01m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a6341c4b61614c4a996d251d23f17cd8 : SUCCESS in 2h 17m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/07c2e0d845bc490b9153973e68dfe4aa : SUCCESS in 45m 11s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/560e5de9e9d8486b8946ff87fb493984 : SUCCESS in 1h 28m 21s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74d5415186a84f37ae6d194fca663b22 : SUCCESS in 1h 59m 14s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/e3dd1a2eb8954ceeb696c53b130211d8 : SUCCESS in 20m 23s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/33bfcd63a62e49048fecc1e97e32533c : SUCCESS in 4m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/545fd35842ad44629a1522db366e1a1f : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":2},{"id":"5cd151a33f971aa0d79f1e8aaca3e1a5adff5749","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-12-06 21:29:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1a4081505ad0e0208fbeec708ab82393cb98f63c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 21:30:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3b73b6d46e07337927c0a9963578f999e51a271a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 23:20:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14ec4c9c75514b56a860f4aee9aaef38\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8fd4c7ca940848b2b8d68413d2a56724 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5141b4aae94642c4a57fbfad0a68b6a3 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0acdb676f70e40acb85137534a279e0d : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0801664208504104adeb60993e435353 : SUCCESS in 5m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd006a9d274b48f5a1efb9e810429e8b : SUCCESS in 6m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6cd38afcb5041c0bdd3ee9edc093bbc : SUCCESS in 1h 38m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/186fcb602539405b92375a10b18ae657 : SUCCESS in 3m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/be143dc24ade458b96bf5c3c28918ce6 : SUCCESS in 1h 47m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0018d7676afd4b95870693280eed551b : SUCCESS in 1h 49m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eb8e6c47930b479980cd5710799ef56f : SUCCESS in 1h 49m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cc1e8860c1ad464382e127dd870f64c9 : SUCCESS in 59m 08s","accounts_in_message":[],"_revision_number":2},{"id":"2fec1d6536480bb9065c2a70cb51d76b63170331","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 23:20:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"96c5bb37941d32f05a2369d0ca30a43675eb60f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 23:22:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6354da8650cd4ba79ae92e33d2230292\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/98205cbd0e004fe0b617cf8bb12b6913 : SUCCESS in 47s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ee0d2e351a644278e617b57ded2febe : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1b3cd2fe54b461d80da67a4a151bf4b : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2},{"id":"08ebf948b5dc55a576599390d6ba3ccc14f087ef","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-12-18 22:16:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"781827607d671bcf5a894586f83e6c8f89f3677d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-01-21 11:44:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"246ba28363ec452d54bce836e0826dcff9f0b1f0","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-21 14:32:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8396d75b31ea4ac4622de1b64b77ab8dfb8ac2e1","revisions":{"1bf336763c41895426c276e5c4e79bab002868c5":{"kind":"REWORK","_number":1,"created":"2024-12-02 14:54:14.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/41/936841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/936841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/936841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/936841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/936841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/936841/1"}}},"commit":{"parents":[{"commit":"5184e50c1a3e40cc3911f69831a8a166312c729a","subject":"Merge \"Switch to neutron-functional job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5184e50c1a3e40cc3911f69831a8a166312c729a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-12-02 14:51:09.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-12-02 14:54:01.000000000","tz":60},"subject":"Add definition of the \u0027qinq\u0027 api extension","message":"Add definition of the \u0027qinq\u0027 api extension\n\nThis new extension allows to enable QinQ VLAN type for the networks.\n\nRelated-bug: #1915151\nChange-Id: If6436c048bf51d503c4f433a900b67e9945965ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1bf336763c41895426c276e5c4e79bab002868c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1bf336763c41895426c276e5c4e79bab002868c5"}]},"branch":"refs/heads/master"},"8396d75b31ea4ac4622de1b64b77ab8dfb8ac2e1":{"kind":"REWORK","_number":2,"created":"2024-12-05 09:33:45.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/41/936841/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/936841/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/936841/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/936841/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/936841/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/936841/2"}}},"commit":{"parents":[{"commit":"5184e50c1a3e40cc3911f69831a8a166312c729a","subject":"Merge \"Switch to neutron-functional job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5184e50c1a3e40cc3911f69831a8a166312c729a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-12-02 14:51:09.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-12-05 09:33:44.000000000","tz":0},"subject":"Add definition of the \u0027qinq\u0027 api extension","message":"Add definition of the \u0027qinq\u0027 api extension\n\nThis new extension allows to enable QinQ VLAN type for the networks.\n\nRelated-bug: #1915151\nChange-Id: If6436c048bf51d503c4f433a900b67e9945965ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8396d75b31ea4ac4622de1b64b77ab8dfb8ac2e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8396d75b31ea4ac4622de1b64b77ab8dfb8ac2e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
