)]}'
{"id":"openstack%2Fneutron-lib~937543","triplet_id":"openstack%2Fneutron-lib~master~I902aac9a2231c84bd8be2d16730397f1657d8ae5","project":"openstack/neutron-lib","branch":"master","topic":"bug/1915151","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-12-12 17:55:20.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-12-12 10:16:28.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I902aac9a2231c84bd8be2d16730397f1657d8ae5","subject":"Add VlanQinqDriverError exception","status":"MERGED","created":"2024-12-11 15:51:23.000000000","updated":"2024-12-12 17:57:21.000000000","submitted":"2024-12-12 17:55:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937543-bug/1915151","meta_rev_id":"1c97fe6b0025dac8ad34481d757b9cbd9cad2b57","_number":937543,"virtual_id_number":937543,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-12 17:55:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-12-12 09:44:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-12-12 15:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-12-12 10:16:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-12-12 15:25:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-11 18:04:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-12 08:59:15.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-12-12 09:44:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-12-12 15:25:45.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"e9fa33433ede76ac31c2a0394bc6db15653dbfc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-12-11 15:51:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0194cd90338cc2dab539573edb0ac8eadb2d8a67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 18:04:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be0dcba116f64186bce9e0959155b867\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ebfbb2bb58f4965b98ae705f0d2a334 : SUCCESS in 8m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/923ea7b027e44ee69751d0abe8da469f : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db7216b9868c47e380963a081c90333d : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a93468d3263749b4813cc2d4fc5d483e : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/783b8a0eba6f4074afc0b4abc22826f5 : SUCCESS in 4m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28c11b3ff2774425a744050ca85a7fd2 : SUCCESS in 1h 41m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6e1484da550749adbab84c48bb4d23b9 : SUCCESS in 1h 31m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/555041f2524f4385b7cb78ad9a3cbcd4 : SUCCESS in 1h 53m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/388bf6006c144964913c3f2dc6045738 : SUCCESS in 1h 54m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cdf8380d6f5a4b4a8e8258966c79766c : SUCCESS in 36m 07s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2ee12aff7594997a875fe8ff23e2168 : SUCCESS in 1h 44m 17s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e35a047cc074fe89c7760e3d5bb5449 : SUCCESS in 2h 11m 23s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/72959bc8ff21422abcb70e32922698b6 : SUCCESS in 23m 22s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/17c10348079a48128ae44b64c382cd16 : SUCCESS in 7m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f5937e1b06454adb88b6064fa6247a11 : SUCCESS in 45m 10s","accounts_in_message":[],"_revision_number":1},{"id":"58275fb87711c6792582500374197661260b35c7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-12-12 08:59:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5cbfd8c7798a45af3af8a95e8ea13caef505c147","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-12-12 09:41:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0125dfebc0c57cf59fc560ea63746c0ef4dd0881","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-12 09:44:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"11a5bb813e88586e72dee4e9a9e13c77e2dc44a1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-12-12 10:16:28.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a8d900eefd3523d3b7c91f9c519d6368645ad3bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 12:10:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c44d09cc8cf4baf8035d690fd52c172\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1503d0436fdc4f1e9dec92d145827e24 : SUCCESS in 5m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c1c58f151924b5ebe76aec44718ec49 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fd56c8ed3c64cd09778798d1a73f8d6 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c80c16f778e48729a64c80f1a24a7d2 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c83ad9f637764f3b815ca39dfc248abc : SUCCESS in 2m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cd7793e6baa4e19b39551e6e8406bf3 : SUCCESS in 1h 05m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c1843d690a7c43a59f4e6bd951a5056e : SUCCESS in 1h 23m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b5d4f85af61a4fe69fe2c118266f74a0 : SUCCESS in 2h 02m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3c19828f2da54e39ab585a06a4de3f4c : SUCCESS in 2h 01m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/24611985fd534472af0e5d6822fb2a50 : SUCCESS in 41m 16s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a802692b856461fb48b103e44555ab1 : SUCCESS in 1h 29m 55s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0541c7cddd9a47f9b37af2a4d3e2b32c : SUCCESS in 2h 26m 46s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/f3a2aabde62e4d9ca19bfd7db56f9f2e : SUCCESS in 24m 00s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c7bea5c1304049b39587cd8be4cafafe : SUCCESS in 6m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d656511994d84c5c82569f0afe091bb3 : SUCCESS in 45m 38s","accounts_in_message":[],"_revision_number":2},{"id":"5fe832b445d5a24f4c454948382b48823ed97cf8","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-12-12 15:25:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d3c2d8704b3b9fb7cf7edbb40c39864a6bc014af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 15:26:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fde39b7d9ea66b56a23788742e60c6fa9f175cd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 17:55:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec841cf0de174b5c82b4e7d4021a2870\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a9b9692e52a44c69877fb8a2635f98c : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91d3984d5f694980a628f8513449751c : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9db4e7ac1382404b895627d0d906fbf8 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17fb4eaa880649c59c20825f1b97e0ca : SUCCESS in 3m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18f781a256ad480caaec8c3cff83e2f9 : SUCCESS in 1h 41m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/24e4ab9183c04185a6136b5603754824 : SUCCESS in 1h 44m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4d14a272fccc4c6daae5ee5db5ee8899 : SUCCESS in 2h 27m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2821c2a05ae94f428502bf26a4066aec : SUCCESS in 1h 38m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/655e18f863fc4b159db514dc9f677cba : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":2},{"id":"c0a00f20b77b4854120602df91cf819fc5e718e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 17:55:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1c97fe6b0025dac8ad34481d757b9cbd9cad2b57","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 17:57:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1f9ff9c3f93479280305579a18e89c7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84dd3ddc68ca43d9a4c70eeb83c0c27f : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"238e94f11dedc2543241900b1ed7284824d0be8c","revisions":{"2edf409f01a3e81ced0e9dc111554fe0be739b70":{"kind":"REWORK","_number":1,"created":"2024-12-11 15:51:23.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/43/937543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/43/937543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/43/937543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/43/937543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/43/937543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/43/937543/1"}}},"commit":{"parents":[{"commit":"ff665ba6a427b8b60d8d80d5e189b6eac2166623","subject":"Merge \"Add definition of the \u0027qinq\u0027 api extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ff665ba6a427b8b60d8d80d5e189b6eac2166623"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-12-11 15:48:17.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-12-11 15:51:19.000000000","tz":60},"subject":"Add VlanQinqDriverError exception","message":"Add VlanQinqDriverError exception\n\nThis is follow up for the [1] which introduced API definition for the\nnetwork\u0027s ``qinq`` attribute.\n\n[1] https://review.opendev.org/c/openstack/neutron-lib/+/936841\n\nRelated-bug: #1915151\nChange-Id: I902aac9a2231c84bd8be2d16730397f1657d8ae5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2edf409f01a3e81ced0e9dc111554fe0be739b70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2edf409f01a3e81ced0e9dc111554fe0be739b70"}]},"branch":"refs/heads/master"},"238e94f11dedc2543241900b1ed7284824d0be8c":{"kind":"REWORK","_number":2,"created":"2024-12-12 09:41:07.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/43/937543/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/43/937543/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/43/937543/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/43/937543/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/43/937543/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/43/937543/2"}}},"commit":{"parents":[{"commit":"ff665ba6a427b8b60d8d80d5e189b6eac2166623","subject":"Merge \"Add definition of the \u0027qinq\u0027 api extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ff665ba6a427b8b60d8d80d5e189b6eac2166623"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-12-11 15:48:17.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-12-12 09:40:35.000000000","tz":60},"subject":"Add VlanQinqDriverError exception","message":"Add VlanQinqDriverError exception\n\nThis is follow up for the [1] which introduced API definition for the\nnetwork\u0027s ``qinq`` attribute.\n\n[1] https://review.opendev.org/c/openstack/neutron-lib/+/936841\n\nRelated-bug: #1915151\nChange-Id: I902aac9a2231c84bd8be2d16730397f1657d8ae5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/238e94f11dedc2543241900b1ed7284824d0be8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/238e94f11dedc2543241900b1ed7284824d0be8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
