)]}'
{"id":"openstack%2Fneutron-lib~938852","triplet_id":"openstack%2Fneutron-lib~master~I7db4698a247df18cf5fa1ce7f8ede5c5f27ecfa0","project":"openstack/neutron-lib","branch":"master","topic":"is-bsd","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2025-01-10 17:33:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7db4698a247df18cf5fa1ce7f8ede5c5f27ecfa0","subject":"Deprecate is_bsd and remove its use in test suite","status":"MERGED","created":"2025-01-09 22:43:51.000000000","updated":"2025-01-10 17:34:08.000000000","submitted":"2025-01-10 17:33:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938852-is-bsd","meta_rev_id":"e129a4f0b616b731624ba405c58fb369e1357cac","_number":938852,"virtual_id_number":938852,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-10 17:33:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-01-10 08:18:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-01-10 15:11:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-01-10 15:11:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-10 01:20:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 08:18:57.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-01-10 15:11:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"035b36c64fe6e17138370b8d326307ec1ab94887","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-09 22:43:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15fc846a0bdd69c7665409a4617c9494fcc16543","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 01:20:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d7e57140b014ceba886b908d10e70e8\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5df40c7d3ed044ea95c913a5b5819720 : FAILURE in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5249ba1cdee4c28829f78809fac8b99 : SUCCESS in 5m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1adf036600374765b57b5b4bc7fcce48 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eb166e29f06473299089465484284c3 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05717c49d8984377ad47e549669a5650 : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7eca072a3657435ea83b3615d6a8ad51 : SUCCESS in 3m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/19747baf0c354b949cc0491685608cbe : SUCCESS in 1h 44m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dabd842ab7c045d6a7781d33aac5d1f3 : SUCCESS in 2h 06m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b6f8a9a7086642f1ba19bb9108e458c5 : SUCCESS in 2h 19m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a0e47cf7bb5d4641b9743b1a5ca91562 : SUCCESS in 2h 33m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/24d85f9784724d628c7f4cd37890fef3 : SUCCESS in 58m 00s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e81315fb44e4a1a93ad1ccb35af4951 : SUCCESS in 1h 45m 26s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c5dc011c1fd4cf6b9f0cd735bbbeac8 : FAILURE in 1h 57m 36s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/fd4ec03b631648d39f08a350bbe78e3a : SUCCESS in 20m 59s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7c2cdd52a4e24f8085157edf25ac6abe : SUCCESS in 5m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d02a1ffce65d4e23a97bbca268c44ed4 : SUCCESS in 1h 08m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6e6e5958c96b97965c4b55c7072ba7ebc59a24b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-10 01:44:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"65985015539efa434ccf51f4bf9c43acfee42cdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 04:21:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e6e70384d394a1fb9f2ec4c49d316e2\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/36ae74c0b07c4cdda3915abb26858e0d : SUCCESS in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8cfae68f93f49e5ac517f6ba083e3a9 : SUCCESS in 6m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4ab4e386ede4b1daa9a985e501c80b3 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8b2602c1431474e9d81afdcfaa68c8b : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/293f792c94964854b51513e76e60e8e8 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7953acad38324c1db92bee40df8fb85e : SUCCESS in 6m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a086b1f30b24b62889288707f9530b8 : SUCCESS in 1h 39m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2de7a639718e4fb5a43ed409a13e509b : SUCCESS in 1h 50m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d08cb9f067314b8e912536dfe7a55195 : SUCCESS in 2h 36m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/20bcf7051d2b4a8882ff5d79b953bc6a : SUCCESS in 2h 16m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb0264ace4004b5eb0a4ad518db8d3bf : SUCCESS in 57m 48s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5830628136064c229328882426fbbfe4 : SUCCESS in 2h 00m 21s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d95aa640d28471c954e268ab72fada8 : SUCCESS in 1h 24m 29s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/d6866cfa7163459d854f8a973751da7e : SUCCESS in 23m 09s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/b6123e7f6c54486280d6f19750c10fe4 : SUCCESS in 7m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f2e941fcd610455fa9f2ab43e05640a5 : SUCCESS in 1h 24m 51s","accounts_in_message":[],"_revision_number":2},{"id":"5caa09794b0af20130e05d609087cfe8dced28f4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-10 08:18:57.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"402ee34b72d9a435ad1e27ae6fd3303577a2b618","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-01-10 15:11:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2c15a49fc582ab6e4fba19aa5e6d692ee671a565","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 15:11:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5e588e753afeb9ec8ff3204976ccc5e74dd55760","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 17:33:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba58beeb7f3842398f8a2c149c474232\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/936606b7ba1047e2bb962cc4cac815ee : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dff240b9f5643a7a301d20508e50564 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de2ca74a6256456888e6a0cf2039cae4 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bf19e878204429b974dc37603390254 : SUCCESS in 5m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/be3fb16fd76e40089732418b77c8803b : SUCCESS in 3m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c346b55b96342ad9f1460ae5b4de18f : SUCCESS in 1h 38m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8215d93288f549649869d7b1c334fe44 : SUCCESS in 1h 40m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1369858fc5484009ab9ea6d5d79a5e5b : SUCCESS in 2h 19m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e1c12b8c63784514991caf5d6583de73 : SUCCESS in 1h 33m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f0957f3d39ba44a0babce0643f76b99f : SUCCESS in 1h 14m 32s","accounts_in_message":[],"_revision_number":2},{"id":"2d8666e1a2d24cfc05b82081a2d3c15714528278","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 17:33:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e129a4f0b616b731624ba405c58fb369e1357cac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 17:34:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12f8a3a3054045f893cc381f4e77a010\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c0417795b2a471db62826565556cc5e : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4d67703e97492d5886b8dbf5aaa0cfd52b0a5702","revisions":{"a033e4bdb2e7fa1b2514b1fc2eddd25975a44ccc":{"kind":"REWORK","_number":1,"created":"2025-01-09 22:43:51.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/52/938852/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/938852/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/938852/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/938852/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/938852/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/938852/1"}}},"commit":{"parents":[{"commit":"ff665ba6a427b8b60d8d80d5e189b6eac2166623","subject":"Merge \"Add definition of the \u0027qinq\u0027 api extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ff665ba6a427b8b60d8d80d5e189b6eac2166623"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2025-01-09 22:41:54.000000000","tz":-300},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2025-01-09 22:43:08.000000000","tz":-300},"subject":"Deprecate is_bsd and remove its use in test suite","message":"Deprecate is_bsd and remove its use in test suite\n\nWe do not support running neutron on bsd (e.g. macos) systems, so the\nfunction has no reason to exist. The only user is neutron test suite and\nit\u0027s proposed to remove its usage there too:\n\nIf5e06dcf3a323692137d2c7cbb8ca725d9d7d821\n\nNote: This patch reintroduces debtcollector dependency to handle the\ndeprecation process.\n\nChange-Id: I7db4698a247df18cf5fa1ce7f8ede5c5f27ecfa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a033e4bdb2e7fa1b2514b1fc2eddd25975a44ccc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a033e4bdb2e7fa1b2514b1fc2eddd25975a44ccc"}]},"branch":"refs/heads/master"},"4d67703e97492d5886b8dbf5aaa0cfd52b0a5702":{"kind":"REWORK","_number":2,"created":"2025-01-10 01:44:28.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/52/938852/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/938852/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/938852/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/938852/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/938852/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/938852/2"}}},"commit":{"parents":[{"commit":"ff665ba6a427b8b60d8d80d5e189b6eac2166623","subject":"Merge \"Add definition of the \u0027qinq\u0027 api extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ff665ba6a427b8b60d8d80d5e189b6eac2166623"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2025-01-09 22:41:54.000000000","tz":-300},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2025-01-10 01:44:24.000000000","tz":-300},"subject":"Deprecate is_bsd and remove its use in test suite","message":"Deprecate is_bsd and remove its use in test suite\n\nWe do not support running neutron on bsd (e.g. macos) systems, so the\nfunction has no reason to exist. The only user is neutron test suite and\nit\u0027s proposed to remove its usage there too:\n\nIf5e06dcf3a323692137d2c7cbb8ca725d9d7d821\n\nNote: This patch reintroduces debtcollector dependency to handle the\ndeprecation process.\n\nChange-Id: I7db4698a247df18cf5fa1ce7f8ede5c5f27ecfa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4d67703e97492d5886b8dbf5aaa0cfd52b0a5702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4d67703e97492d5886b8dbf5aaa0cfd52b0a5702"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
