)]}'
{"id":"openstack%2Fneutron-lib~938856","triplet_id":"openstack%2Fneutron-lib~master~I740bc58ae57e22c878b52215e72aaec6ebca1299","project":"openstack/neutron-lib","branch":"master","topic":"type_tooling","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2025-01-10 17:50:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I740bc58ae57e22c878b52215e72aaec6ebca1299","subject":"Enable mypy pre-commit hook","status":"MERGED","created":"2025-01-09 23:21:18.000000000","updated":"2025-01-10 17:51:40.000000000","submitted":"2025-01-10 17:50:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938856-type_tooling","meta_rev_id":"dcb4d63d988caca3f00e6e88eaff1c2dd986896c","_number":938856,"virtual_id_number":938856,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-10 17:50:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-01-10 08:06:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-01-10 15:13:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-01-10 15:13:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-10 01:52:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 08:06:17.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-01-10 15:13:39.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"c92be88c861aec7123c9a5911b2ecd2abdb0faf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-09 23:21:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37f289130b0c2c54000eea15e53dc0b67146aeec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 01:52:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b353dd5cda354c769d559439457b660c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/57cb95cce40d4b4f9de1489bbace674b : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5b2a360ef5748eabf38cb06ce927bd8 : SUCCESS in 6m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27195e3637054bc8ba2f7ea892bd3261 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/279d946e70eb47b6ba2095c8907340ba : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c4f6f3b074446799a0473fd9f7bf7df : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f54765dbfa514cdfabbf9e958af70d53 : SUCCESS in 5m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/707781c64fc946beadb39214fc39b84e : SUCCESS in 1h 39m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da214693646f486a96c45eb92bd6a238 : SUCCESS in 4m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c145b3aded4f4f819ce1628670c4060f : SUCCESS in 2h 07m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e5be6cc03a3946078a8318f3ed829707 : SUCCESS in 2h 11m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ff1e8f2fb5a74256bd5932643919fdad : SUCCESS in 2h 03m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ad1743418e5a42d89cfa3ebd75843582 : SUCCESS in 44m 44s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3a2d4e7b74f4c7a8f5e4556de924e2e : SUCCESS in 1h 20m 42s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9111aeb64bf645e4af9b040eda1820bd : SUCCESS in 2h 21m 42s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/c673e6d2ac2f44cca157b82ab1a1607f : SUCCESS in 20m 44s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/27b5d51e58144fec8379b7620d27682d : SUCCESS in 5m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3013cfc1a0e04a0ba885c735e476d9b8 : SUCCESS in 1h 17m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5992431df463f1eb14350411adb73205a6665459","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-10 08:06:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0f9a994fcff65c1c931cc505dc8d95b4f5d65054","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-01-10 15:13:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d021970163fc90c3a37f1be9c9dccf84bda5078b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 15:14:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9aeb1e38f2ff489c93661126819ce65cfc9581bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 17:50:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b8c2da855344e22ba63925d9756a103\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a095abe7014949b3a00a31e69d8e0c62 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/113c7ad882674dca961420df2ac0d581 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fa1ed05b7d742cea64160cedc04de77 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/244f44d733374bd784a94128f048c69b : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c405c20ad244bbb805fa08448a553a7 : SUCCESS in 4m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1343366cc1234742b56d8a6ad0ff0be3 : SUCCESS in 1h 34m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee6d6dedfc8b4e86b55285856d776d70 : SUCCESS in 4m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2dd9f3bcfdff4399a5dd606d56b826af : SUCCESS in 1h 34m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/95f18af0b5894dacbbde1314139c5e8c : SUCCESS in 2h 21m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6c3bc45de5f44a21a413759b70bb2f20 : SUCCESS in 2h 33m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ab8bf6ce20324b80bd90c5c988aac90e : SUCCESS in 55m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f2db08f0a34d22d7e3277ba7a17bdb34d0732ae6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 17:50:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dcb4d63d988caca3f00e6e88eaff1c2dd986896c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 17:51:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daa0655aaa7e474f9cd80a4c3bc336f8\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/7c47ed0b8ce84e1894112102d7df576f : SUCCESS in 1m 05s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1808ccbb705c4fe7a38d9bfe517f087a : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a0f58572f1240728ffb2b25e3ff44b2 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1de9764ae9fb5f757b18df4269b048aacac95a8d","revisions":{"1de9764ae9fb5f757b18df4269b048aacac95a8d":{"kind":"REWORK","_number":1,"created":"2025-01-09 23:21:18.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/56/938856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/56/938856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/938856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/938856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/938856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/56/938856/1"}}},"commit":{"parents":[{"commit":"3f59155effc8242eaec111f12e0653d7f8ecfdf8","subject":"Enable mypy for pep8 tox target","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3f59155effc8242eaec111f12e0653d7f8ecfdf8"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2025-01-09 23:20:40.000000000","tz":-300},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2025-01-09 23:20:40.000000000","tz":-300},"subject":"Enable mypy pre-commit hook","message":"Enable mypy pre-commit hook\n\nChange-Id: I740bc58ae57e22c878b52215e72aaec6ebca1299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1de9764ae9fb5f757b18df4269b048aacac95a8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1de9764ae9fb5f757b18df4269b048aacac95a8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
