)]}'
{"id":"openstack%2Fneutron-lib~938857","triplet_id":"openstack%2Fneutron-lib~master~I85076d5171a4bd1c21a3db6cfb8fbe18974f9805","project":"openstack/neutron-lib","branch":"master","topic":"type_tooling","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2025-01-10 17:50:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I85076d5171a4bd1c21a3db6cfb8fbe18974f9805","subject":"Bump mypy to 1.14.1 (the latest)","status":"MERGED","created":"2025-01-09 23:21:18.000000000","updated":"2025-01-10 17:53:05.000000000","submitted":"2025-01-10 17:50:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938857-type_tooling","meta_rev_id":"8ff68dce3ae4bb37fffef2dc44d5af418cba4976","_number":938857,"virtual_id_number":938857,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-10 17:50:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-01-10 08:06:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-01-10 15:13:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-01-10 15:13:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-10 01:00:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 08:06:23.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-01-10 15:13:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"bd5b544b3e5682dea623c7e2e8c6b4696596d695","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-09 23:21:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"819098e9347e4dcea76bab4e9243b0725d324752","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 01:00:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc35c9acd38547bd816fcb47974678a3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/697bd341faaa44eb81600719d1a91fc3 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67b25d92b28e4356aa21768028e5cdd5 : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca837be65b8446119bfef05d56569be8 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3decd8b1bd4241618a31d1d43210b81d : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0818b1f04b2c4a06b35628b74abc7712 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4227b564b1b46eaafe26022f34c99d7 : SUCCESS in 4m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e0a8e7bb63047e0bf641d59ec6456c8 : SUCCESS in 1h 38m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c93c5645719a4176a972c93e81e89b02 : SUCCESS in 3m 47s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/4b58f6962ffa404cb9bc815f07211b73 : SUCCESS in 20m 21s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/dd0ff7bd2d0a485c8f20c67d940e4e93 : SUCCESS in 5m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e9e2e9cf94ba464792582109e4b9e46b : SUCCESS in 1h 17m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1b5b96f8be5addbf6e62de05f2cb65ea5de99fbc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-10 08:06:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"63db22cb49aa11c5ebf15fd395ff84d73a26e2af","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-01-10 15:13:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d174086a94e64b72fa02d724daa76c187cec4d56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 15:14:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a44188fb34d6516c42cd0889ecaa283e5f85e03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 17:50:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/470e7121c3584dbd9890dcc857139f42\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d96cd674746d4f59ba907e9f04cab93d : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27902ca2d99546bb8fe39f17e65e5485 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b48308d315e490098f62d20d3c6f394 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2b373a62964496cb7c6305c721ffc7d : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8b8f3eb38ff4384bdb439c7c594b4ee : SUCCESS in 3m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b25bd89f86fe4bbc8c567990584b6642 : SUCCESS in 1h 37m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efe2d252243445bd8a9d9b19e36021b5 : SUCCESS in 4m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cb48cfe080244da8bc71553232a4a332 : SUCCESS in 1h 13m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5bc3a20c33cd29bcc8e1b177502eb941e1495ceb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 17:50:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8ff68dce3ae4bb37fffef2dc44d5af418cba4976","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 17:53:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e645cd432cf645d1a548660eec03fb96\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6e1eb631bbbf4cc4b9e558c75d276e1e : SUCCESS in 1m 01s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/760d7247a0514453ac84d85e5d5e9860 : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e37f72277424f8081bc86dca437bff5 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"71878575b5273ec925c92c372928c5c38ec5e2b1","revisions":{"71878575b5273ec925c92c372928c5c38ec5e2b1":{"kind":"REWORK","_number":1,"created":"2025-01-09 23:21:18.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/57/938857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/57/938857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/938857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/938857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/938857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/57/938857/1"}}},"commit":{"parents":[{"commit":"1de9764ae9fb5f757b18df4269b048aacac95a8d","subject":"Enable mypy pre-commit hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1de9764ae9fb5f757b18df4269b048aacac95a8d"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2025-01-09 23:20:53.000000000","tz":-300},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2025-01-09 23:20:53.000000000","tz":-300},"subject":"Bump mypy to 1.14.1 (the latest)","message":"Bump mypy to 1.14.1 (the latest)\n\nChange-Id: I85076d5171a4bd1c21a3db6cfb8fbe18974f9805\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/71878575b5273ec925c92c372928c5c38ec5e2b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/71878575b5273ec925c92c372928c5c38ec5e2b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
