)]}'
{"id":"openstack%2Fneutron-lib~940682","triplet_id":"openstack%2Fneutron-lib~master~Ida5afffb808597aa09cc62984b36189a8bf409a0","project":"openstack/neutron-lib","branch":"master","topic":"bug/2093258","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-02-05 10:45:26.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-02-05 10:45:26.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-02-05 10:45:26.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-02-04 16:19:12.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ida5afffb808597aa09cc62984b36189a8bf409a0","subject":"Revert \"Change the ``StandardAttribute`` load method to \"selectin\"\"","status":"MERGED","created":"2025-02-04 09:49:14.000000000","updated":"2025-02-05 10:46:27.000000000","submitted":"2025-02-05 10:45:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":923926,"submission_id":"940682-bug/2093258","meta_rev_id":"00a6307a289c69112242fc5ca24a2f26e10e0301","_number":940682,"virtual_id_number":940682,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-05 10:45:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-02-05 08:16:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-02-04 16:19:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-02-05 08:16:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-02-04 09:52:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-04 09:49:14.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-02-04 09:49:14.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2025-02-04 09:49:14.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-02-04 09:49:14.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-02-04 09:49:14.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-05 08:16:43.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"71bab0ed715f92ad4ff5fd16284fee824c935c80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-04 09:49:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31edc49af0f2e137008d5ded6be61d9779ca86d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 09:49:45.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron-lib for 940682,1","accounts_in_message":[],"_revision_number":1},{"id":"a80d4f736eeae1a28ee2ef5866896bbe4788efbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-04 09:51:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d3a84d228f1097a606b3b48a17e921eb1e83536f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-04 09:52:02.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"c638e15cf1747e9ba4cc88dff5c427469647560f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 11:58:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4faba7ff602340c2a607fdfb4ee2b252\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e366b41ad254531aa432748db961ff8 : SUCCESS in 5m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/733a5ecba0d04fdc9dfd2d080622c4cb : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67fb17e7cfbf41c6b82b42612843bec2 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86c32d604f7b4736816c770440e0861a : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97a48bfb9b3f45abb6ff8a64b081234a : SUCCESS in 4m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4524d2abc075477c9bb4c3169b916123 : SUCCESS in 1h 52m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8d8f75e2fbf74c208c808eebe9a2436b : SUCCESS in 1h 20m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c5a1cf08f9a04cdc88bc73fe2231b061 : SUCCESS in 2h 01m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/edb59193c06f415c9a961ad055088c90 : SUCCESS in 2h 04m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e10a60e0a47740fa8f2a78e1e466daf2 : SUCCESS in 29m 25s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a8be845bd0542d6b3986c5a535a8e6e : SUCCESS in 1h 34m 30s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c0ab5eccb644b6cb4b04b1f20943ecf : SUCCESS in 1h 41m 26s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/0c180923269a44a6b1c8f50ceda17fe7 : SUCCESS in 21m 45s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/af1c5f8a0a934d4fbbcf6000badb23c1 : SUCCESS in 5m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/65f91028a45f443da07d57b2c5afa67e : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":2},{"id":"8659d7ffea14ea4f6783a5fc8962c5f47407d63c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-02-04 16:19:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e10b3f227cb88fbdcd639ba990eaea42b163e466","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-05 08:16:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"37ab0f363bb1d32471be640944da3c542ed6859f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 08:17:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e6422f827b978166f1e257a7508358f317a0032a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 10:45:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/058443ed52aa4bf8ad8eadb5d6056d96\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d260c9c5bd254c608752bf2c3d5bd8e5 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1603793b720843b3b754c23f71c0aa85 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/443a68d0fdce4951b5e0433563375078 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/883682747fa04eef958ce1744d4b4f27 : SUCCESS in 5m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06898f36930d4ac78caff18d01b00c21 : SUCCESS in 1h 41m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8a4cd11c12e44419b7f4244e03eed527 : SUCCESS in 1h 56m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2df9b26d29e5410484b04eaeb107fc50 : SUCCESS in 2h 20m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d5354740a7eb42b89658f18e682c453d : SUCCESS in 1h 43m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/704ffe06ef9740f1b69847f8ff0fca73 : SUCCESS in 1h 06m 36s","accounts_in_message":[],"_revision_number":2},{"id":"7f993856fe775769a82952d1432196bfcda6188e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 10:45:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"00a6307a289c69112242fc5ca24a2f26e10e0301","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 10:46:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a5d7b62b3204fbbba6a2272b35df160\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cab3488d2d9243f78898cc1d8aa3edc8 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e2d68e39f5e62be9ca8a247b3f957e5e2d4ec3a3","revisions":{"eb7216870edd0127f92297e02a4b9bf2a0606255":{"kind":"REWORK","_number":1,"created":"2025-02-04 09:49:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/82/940682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/82/940682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/82/940682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/82/940682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/82/940682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/82/940682/1"}}},"commit":{"parents":[{"commit":"7f702af482b6b47f2fe74a8b475fb2a23cb5f7a4","subject":"Change the ``StandardAttribute`` load method to \"selectin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7f702af482b6b47f2fe74a8b475fb2a23cb5f7a4"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-02-04 09:49:14.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2025-02-04 09:49:14.000000000","tz":0},"subject":"Revert \"Change the ``StandardAttribute`` load method to \"selectin\"\"","message":"Revert \"Change the ``StandardAttribute`` load method to \"selectin\"\"\n\nThis reverts commit 7f702af482b6b47f2fe74a8b475fb2a23cb5f7a4.\n\nReason for revert: this patch was introduced to make the database\nproduced code compatible with PostgreSQL. During this cycle we dropped\nthe support for this backend. This patch is probably introducing a race condition when retrieving the port standard attributes because these are retrieved in a second SQL query (lazy\u003d\u0027selectin\u0027).\n\nIn order to keep a healthy database engine, supporting MariaDB/MySQL, it is better to retrieve the selected field in the ``port`` retrieval query instead of issuing a delayed second one.\n\nChange-Id: Ida5afffb808597aa09cc62984b36189a8bf409a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/eb7216870edd0127f92297e02a4b9bf2a0606255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/eb7216870edd0127f92297e02a4b9bf2a0606255"}]},"branch":"refs/heads/master"},"e2d68e39f5e62be9ca8a247b3f957e5e2d4ec3a3":{"kind":"REWORK","_number":2,"created":"2025-02-04 09:51:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/82/940682/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/82/940682/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/82/940682/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/82/940682/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/82/940682/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/82/940682/2"}}},"commit":{"parents":[{"commit":"4b30a94c883ebddeb38b097ca050e472a0e6d487","subject":"[API-REF] Add clarification about QinQ and transparent vlan differences","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4b30a94c883ebddeb38b097ca050e472a0e6d487"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-02-04 09:49:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-02-04 09:50:20.000000000","tz":0},"subject":"Revert \"Change the ``StandardAttribute`` load method to \"selectin\"\"","message":"Revert \"Change the ``StandardAttribute`` load method to \"selectin\"\"\n\nThis reverts commit 7f702af482b6b47f2fe74a8b475fb2a23cb5f7a4.\n\nReason for revert: this patch was introduced to make the database\nproduced code compatible with PostgreSQL. During this cycle we\ndropped the support for this backend. This patch is probably\nintroducing a race condition when retrieving the port standard\nattributes because these are retrieved in a second SQL query\n(lazy\u003d\u0027selectin\u0027).\n\nIn order to keep a healthy database engine, supporting\nMariaDB/MySQL, it is better to retrieve the selected field in the\n``port`` retrieval query instead of issuing a delayed second one.\n\nCloses-Bug: #2093258\nChange-Id: Ida5afffb808597aa09cc62984b36189a8bf409a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e2d68e39f5e62be9ca8a247b3f957e5e2d4ec3a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e2d68e39f5e62be9ca8a247b3f957e5e2d4ec3a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
