)]}'
{"id":"openstack%2Fneutron-lib~941895","triplet_id":"openstack%2Fneutron-lib~master~I0f99d0306c580ef77c59d2d27249190883f93bff","project":"openstack/neutron-lib","branch":"master","topic":"taas_api_enums","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-03-03 11:23:33.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-03-03 15:54:13.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-03-03 14:29:50.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I0f99d0306c580ef77c59d2d27249190883f93bff","subject":"taas: Split direction enum","status":"MERGED","created":"2025-02-17 10:47:56.000000000","updated":"2025-03-03 16:32:01.000000000","submitted":"2025-03-03 16:30:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"941895-taas_api_enums","meta_rev_id":"1bb9d299e28d15f64cbf7ec752ba8c0220ca6e65","_number":941895,"virtual_id_number":941895,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-03 16:30:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-03-03 11:23:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-03-03 14:29:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-03 15:54:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-03-03 14:29:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-17 13:09:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-18 19:59:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-02-25 14:08:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-03-03 15:54:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"dbf3de8bf3c9804b45748a1967e1ab20439e3058","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-17 10:47:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2e14d702134ffd133cb57b482785a19d3257b45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 13:09:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f78112f19be242fa94f7b211e94f1194\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4379284a486f49bc99392fa4e4175dc4 : SUCCESS in 7m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccacd45c01c6410cb0c6473ef6cda0c4 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4dcc38d6f5948409ae04fc39c93e8b9 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73f9a7bfcfdd4779bf95d1899500b66e : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c78652983b96430cb04a9643b2801752 : SUCCESS in 3m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/919731930b264f2c8262dd3974778fcf : SUCCESS in 1h 29m 58s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/581930bd832c4b9e9709bc3391763240 : SUCCESS in 2h 18m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7a5f321078444c1d86845a857e8c22a1 : SUCCESS in 1h 48m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/69510b8d3a974a27bd22d28c5b9cd7f3 : SUCCESS in 1h 41m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/147db09bc547459f904fd95613a3b171 : FAILURE in 14m 22s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/101189176591484a9fb2b9ef07a5df56 : SUCCESS in 1h 22m 57s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54b48140414b44e5aa4529dce2ba0183 : SUCCESS in 1h 57m 55s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/08056bce76dc4ed8b5a3929e97ac452d : SUCCESS in 20m 54s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/91bfcb7803c041b8be183f3ade92b4b7 : SUCCESS in 5m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f7f12ac37d1c49e2bbe0e65c48437fe7 : SUCCESS in 1h 14m 03s","accounts_in_message":[],"_revision_number":1},{"id":"67f4347ddef87a295fe280ad69e99f1a26e93344","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-02-18 19:59:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c6373306c16dc161b8370ae68c524ecfe3d44841","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-25 14:08:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"980d84ee5e0f19ab036de9656eb73cfc9cce8e6c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-02-25 15:58:42.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dbc59180744d38373dcfb52b9d593dff900aa02a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-03-03 10:52:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7a3ae381e106e39f6d0bb06da0ba2ff8abdf9896","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-03-03 10:53:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"95d5fbe35d98b07ddcb111274131b2a05ba1be66","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-03 11:23:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8f5808cd04a03e4baee4f79c96b94890f7540ff6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 13:11:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51da5d9f434342b4bcf28a819b57f7e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c9a27c752cb4398974281e309fdb3c5 : SUCCESS in 6m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b17315cd1991468d9dd3948541622be1 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c04421c120c04da581a60db85643d7e4 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6b7a561f3e1486cba7569beb849cac0 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26c62f12ecf04dd681d11476efbc93cf : SUCCESS in 4m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/255e9d79f0e249aea577b035515e5600 : SUCCESS in 1h 46m 13s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/26bc7c8a29c845d298a549f7751fd616 : SUCCESS in 1h 55m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d1a50d725dd3476bae709c93ca7d856d : SUCCESS in 1h 58m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bb8919cf2cef4972a938b695cdd7b089 : SUCCESS in 41m 20s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1533bb87ffc74f79930a819d098fb44c : SUCCESS in 1h 58m 33s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5eeb36a507e54674ab94d2cffcbc7bae : SUCCESS in 2h 17m 42s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/d8503dbafc4943ae864abedc1a9c8788 : SUCCESS in 22m 42s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/4fc796006c7b4c91915c47266af57c9b : SUCCESS in 5m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0a065f0b8c1a499ea2f8c75cd1dd69d1 : SUCCESS in 1h 17m 06s","accounts_in_message":[],"_revision_number":2},{"id":"c71e2f5c1cae1ddc7386ba8764098398bc36ccec","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-03-03 14:29:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fdd2dc1bc4a65bad7b7801a48cb7d372938d7be8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 14:30:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4edf8d0ce6265c5d8e745e5007d2d6a78f201a09","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-03 15:54:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1749333ceee7c8a70d143915c328c89ab9b21ebf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 16:30:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8576e9c046324beabc1760e3485b5e8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3fc570276b24247b5eac42df1458580 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dc49f7b1df044d7a347571be0da444d : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4ec9094ba4e41aa9cdd5832aabe046f : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f274eb2e9ec64fd99b7163be44e23b56 : SUCCESS in 5m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/775b74fd41534b9fb01e59698b2a8fa6 : SUCCESS in 1h 39m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/229086ec6203470aabc80700c2bb2eb8 : SUCCESS in 1h 55m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f701f01fbb954f6b8c5bc5bf82785288 : SUCCESS in 1h 59m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a375acc732b6490888886d2b1de1a7ba : SUCCESS in 1h 30m 26s","accounts_in_message":[],"_revision_number":2},{"id":"1d5b83918a5dc485d965a0fad1f0a5e869fdb317","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 16:30:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1bb9d299e28d15f64cbf7ec752ba8c0220ca6e65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 16:32:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73045f7823bf4728b8c3a9cc2066337b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74afc0da514e4eb8b3d9280a81b46c5c : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2e959fd7f1f6becab862113acdb0cd9d60b53552","revisions":{"5d3ea91a41245e4ee23f22eecca4a20abe2f9980":{"kind":"REWORK","_number":1,"created":"2025-02-17 10:47:56.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/941895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/95/941895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/95/941895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/95/941895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/95/941895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/95/941895/1"}}},"commit":{"parents":[{"commit":"e2d68e39f5e62be9ca8a247b3f957e5e2d4ec3a3","subject":"Revert \"Change the ``StandardAttribute`` load method to \"selectin\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e2d68e39f5e62be9ca8a247b3f957e5e2d4ec3a3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-17 10:36:33.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-17 10:36:33.000000000","tz":60},"subject":"taas: Split direction enum","message":"taas: Split direction enum\n\nThe direction enum for taas (IN, OUT, BOTH) used\nin many places in tap-as-a-service, split it to make it\neasier\n\nChange-Id: I0f99d0306c580ef77c59d2d27249190883f93bff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5d3ea91a41245e4ee23f22eecca4a20abe2f9980"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5d3ea91a41245e4ee23f22eecca4a20abe2f9980"}]},"branch":"refs/heads/master"},"2e959fd7f1f6becab862113acdb0cd9d60b53552":{"kind":"REWORK","_number":2,"created":"2025-03-03 10:52:33.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/941895/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/95/941895/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/95/941895/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/95/941895/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/95/941895/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/95/941895/2"}}},"commit":{"parents":[{"commit":"e2d68e39f5e62be9ca8a247b3f957e5e2d4ec3a3","subject":"Revert \"Change the ``StandardAttribute`` load method to \"selectin\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e2d68e39f5e62be9ca8a247b3f957e5e2d4ec3a3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-17 10:36:33.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-03-03 10:52:23.000000000","tz":60},"subject":"taas: Split direction enum","message":"taas: Split direction enum\n\nThe direction enum for taas (IN, OUT, BOTH) used\nin many places in tap-as-a-service, split it to make it\neasier\n\nChange-Id: I0f99d0306c580ef77c59d2d27249190883f93bff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2e959fd7f1f6becab862113acdb0cd9d60b53552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2e959fd7f1f6becab862113acdb0cd9d60b53552"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
