)]}'
{"id":"openstack%2Fneutron-lib~952978","triplet_id":"openstack%2Fneutron-lib~master~I3180b1d88ee5dfaa7cc00758ed65a894d3691597","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-06-26 10:49:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3180b1d88ee5dfaa7cc00758ed65a894d3691597","subject":"Prepare migration of NetworkMTUSubnetConflict","status":"MERGED","created":"2025-06-20 12:51:50.000000000","updated":"2025-06-26 10:50:13.000000000","submitted":"2025-06-26 10:49:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952978","meta_rev_id":"b7318945682a99afcc67c0f3206d6aa3a44b9ad1","_number":952978,"virtual_id_number":952978,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-26 10:49:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-06-20 13:31:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-06-23 11:43:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-06-26 08:38:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-20 13:31:14.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-06-20 15:22:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-23 11:43:53.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"99730d9552b76e1817dcb204849ede88b63da10f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-20 12:51:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a65e7a95e32015ced986eaacfc4a45b5fb5f1bfe","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-20 13:31:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5dede0131e024851f0a31640243ede1de7257a23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 15:22:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6f5bda22a924299a4bdf5bf6dab81d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/086ed4579f864275a7e2868408f35910 : SUCCESS in 9m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61539370ecc64daba544826176e0248d : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0282277233742f1a52032aba77207bc : SUCCESS in 7m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a679b63cb69c4686932c598f90ec2a82 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd865f83e59e445c813e4250d7ae3076 : SUCCESS in 5m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a52dfbc1a73f4f6cac63cd4ea6a97791 : SUCCESS in 7m 48s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1a9bd039440424b86f927eda54b36f7 : FAILURE in 1h 44m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b05e00ce41404d9b9a15d234919ea136 : SUCCESS in 1h 27m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/caf811ce6a10478b99c1fef7397a4cd7 : SUCCESS in 2h 28m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/236dc73813194022bdc817e639b27542 : SUCCESS in 52m 16s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/c71f2c08feff47ca9ebccde38ee527a1 : SUCCESS in 22m 01s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0b54204ae92549cab9a0038b0bc9f1f9 : SUCCESS in 7m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f9c51bd98569425991891dc8690d7996 : SUCCESS in 1h 35m 59s","accounts_in_message":[],"_revision_number":1},{"id":"88a2c4cdb691db79a970b7a55491ee931dfdec9f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-23 11:43:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a10b269ff3a2d328765fd8a5d4e1977ff28ac8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 14:16:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31ba6e8f709445f082277faaace30976\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/431e4140c6774a10a4c409a4d4b26eca : SUCCESS in 10m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e94f72d895394e0180502e29b100f72f : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b216146b9724aa88d3097580dff4dfb : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55f403ba02f6490c8684d6a801044a84 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c36e0ca5d17441d9b0f9c8012d116e73 : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ebfe6701667b4ed29c4913c9d2668d8e : SUCCESS in 6m 04s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3430570a5611417d98f3a1c7ba453a19 : SUCCESS in 1h 24m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bf5b74fe247a47c6a1bb1edaea8f0111 : SUCCESS in 1h 28m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b0b0d92f410e48b990c45f5a218010e4 : SUCCESS in 2h 31m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1acf2495bfa140f18eff3c72aee9c7e3 : SUCCESS in 41m 58s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/41bfcd21fdd14673bd43ef8ce10d7140 : SUCCESS in 19m 33s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ac4cc4ef8b7a4618a6bf17b5e18dafd8 : SUCCESS in 4m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c20e2cdf1255462a80c6619ca0ca062a : SUCCESS in 1h 38m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1266240eaac691c53126983774805e33352fd2ed","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-26 08:38:35.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9cc3d4c6473be92843ec91d67c7727be30d8deac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 08:38:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"584639c2fd84c77fa4f7b7559d9a4f48eb871ee0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 10:49:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3751685adff04504a641adcea8843c44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2185cb0e356942bf8a991d16bdaaa12c : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5baee071f513467095a8ff3b072316e3 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5183ff0a41194974ab8d922702de0f6a : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b70a1453ec90477a9a470822d9f481b4 : SUCCESS in 3m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1930e025b14e4628ba84748c7818a780 : SUCCESS in 1h 57m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3086f09d9ae04fa2a6b7213326189b38 : SUCCESS in 1h 21m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/203d033361594aeeadd9ed89891b83e7 : SUCCESS in 2h 09m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b05fd486673e467e8379e992f03ad538 : SUCCESS in 1h 21m 37s","accounts_in_message":[],"_revision_number":1},{"id":"43cce2f0233d771a9f974994bc2270f683f9e054","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 10:49:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b7318945682a99afcc67c0f3206d6aa3a44b9ad1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 10:50:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7bac17709d34904819ad81692624d35\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03364c1d44b1499280e8576067b6fa43 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"48b8f7b26482b3df0bf6df3019110d51a8844558","revisions":{"48b8f7b26482b3df0bf6df3019110d51a8844558":{"kind":"REWORK","_number":1,"created":"2025-06-20 12:51:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/952978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/78/952978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/952978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/952978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/952978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/78/952978/1"}}},"commit":{"parents":[{"commit":"094d102b5f728e1f9789e49c2fb97c36376f46f2","subject":"Merge \"tests: Stop using ddt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/094d102b5f728e1f9789e49c2fb97c36376f46f2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-20 12:51:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-20 12:51:02.000000000","tz":540},"subject":"Prepare migration of NetworkMTUSubnetConflict","message":"Prepare migration of NetworkMTUSubnetConflict\n\nIt was added to neutron by 88ce859b568248a0ee2f47a5d91c1708b774d20e ,\nbut was supposed to be moved to neutron-lib later.\n\nChange-Id: I3180b1d88ee5dfaa7cc00758ed65a894d3691597\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/48b8f7b26482b3df0bf6df3019110d51a8844558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/48b8f7b26482b3df0bf6df3019110d51a8844558"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
