)]}'
{"id":"openstack%2Fneutron-lib~954444","triplet_id":"openstack%2Fneutron-lib~master~I4b6f742d2d1b169eedb81799eac24987315fa7ea","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-07-09 15:39:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4b6f742d2d1b169eedb81799eac24987315fa7ea","subject":"Add ALLOWED_ADDRESS_PAIR to the callbacks resources","status":"MERGED","created":"2025-07-09 10:50:12.000000000","updated":"2025-07-09 15:40:36.000000000","submitted":"2025-07-09 15:39:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954444","meta_rev_id":"076113b2a89cff8539d091cbc433fcbdde77c1ae","_number":954444,"virtual_id_number":954444,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-09 15:39:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-07-09 13:35:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-07-09 13:32:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-07-09 13:35:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-09 13:10:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-09 13:32:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-07-09 13:35:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"b04af2db04e851713c2ccda9884b6e5b20e650e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-07-09 10:50:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c1d9e580fafd17f7c63acdf05350cc849244b0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 13:10:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e26d9d5495040478813a61e874e09f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f06a4952245e460c9e5ff6383844a3c6 : SUCCESS in 8m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbd9dfa27b104e94a7ee59cc98d27bcb : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad0dcec99570488c909937aaa3880638 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/774cd0510976430aa4942159980ce681 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/42038285459e4af5bfae492fe01b1c85 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8a892bc9bbd4149afbbaf45b94755f0 : SUCCESS in 6m 06s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34614a32e58440db910040d6f596ad45 : SUCCESS in 2h 00m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6461d0d196d8483c8a6ac377a7ba1226 : SUCCESS in 1h 23m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/25d7856293eb41eb8c33088a036754f2 : SUCCESS in 2h 19m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/99b3e641970140d2820b1d3449323327 : SUCCESS in 34m 56s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/c87d9bc6e10b4bd1a4eb13ec77edb69a : SUCCESS in 22m 47s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/cf4f6a45017642af88448caa74fa2f04 : SUCCESS in 5m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/493708bc3ab54b5db93c202b832c96d4 : SUCCESS in 1h 38m 52s","accounts_in_message":[],"_revision_number":1},{"id":"156d0d7bc8751ffb987f863f28adb235bb3630e1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-09 13:32:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ea2d736b66a6bb4975a93ef9217199fd8c8b72de","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-07-09 13:35:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f81f3297d214e38afe9e3c3d381db56e9351401","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 13:36:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e8742e856f1bc5777d2f37e8dbd5fcfb673b8b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 15:39:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/334176b716574fb08e9300ce4bd1dc5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ede88a20bb1f44dca8342cf174981fd2 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3af4ac63c32947deb047387000c7f300 : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0284197de5774a27a4f8d3505d107f05 : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/edf9aea208a4457a81b22c0e409c715a : SUCCESS in 2m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76e89af2c0b74f70bd75f8c6cb36ba94 : SUCCESS in 1h 21m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3e06030b8827401c8287ad5c4b1af67b : SUCCESS in 1h 40m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/86e4b185a08f4ebcb5a9d033278f4950 : SUCCESS in 2h 02m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/59eace4515234127946159ef1162316b : SUCCESS in 1h 38m 49s","accounts_in_message":[],"_revision_number":1},{"id":"27f77ec1005611fc7cc9ac70220ecfa72a0d9f35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 15:39:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"076113b2a89cff8539d091cbc433fcbdde77c1ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 15:40:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dd5685c532f4090a7d00cdd9764a1fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d3079cb30f74983b3ef9f10ff829904 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e21c58f13f650a460a305af54e173238ec7f5cd7","revisions":{"e21c58f13f650a460a305af54e173238ec7f5cd7":{"kind":"REWORK","_number":1,"created":"2025-07-09 10:50:12.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/44/954444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/44/954444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/44/954444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/44/954444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/44/954444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/44/954444/1"}}},"commit":{"parents":[{"commit":"b49ef80fe02fbd36afd5a301dbe1eca236a1f5d0","subject":"Merge \"sqlalchemy: Use built-in declarative\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b49ef80fe02fbd36afd5a301dbe1eca236a1f5d0"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-07-09 10:49:05.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-07-09 10:49:05.000000000","tz":120},"subject":"Add ALLOWED_ADDRESS_PAIR to the callbacks resources","message":"Add ALLOWED_ADDRESS_PAIR to the callbacks resources\n\nRelated-bug: #2116249\n\nChange-Id: I4b6f742d2d1b169eedb81799eac24987315fa7ea\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e21c58f13f650a460a305af54e173238ec7f5cd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e21c58f13f650a460a305af54e173238ec7f5cd7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
