)]}'
{"id":"openstack%2Fneutron-lib~958490","triplet_id":"openstack%2Fneutron-lib~master~I7fecb3f34b4b3046587c1a1bc28f7dc7736907eb","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-08-26 10:29:57.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-08-26 08:19:27.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-08-26 10:29:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7fecb3f34b4b3046587c1a1bc28f7dc7736907eb","subject":"Replace \"tenant\" with \"project\"","status":"MERGED","created":"2025-08-25 20:54:25.000000000","updated":"2025-08-26 10:32:53.000000000","submitted":"2025-08-26 10:29:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958490","meta_rev_id":"b2630bb9399a129edec3b8d7c69aba314b5fc142","_number":958490,"virtual_id_number":958490,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-26 10:29:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2025-08-26 08:02:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-08-26 08:19:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2025-08-26 08:19:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-25 23:21:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-26 07:42:35.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-08-26 07:42:45.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-08-26 08:02:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"ac8288557751cfbedf8655d86c35b5726b99c0d4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-08-25 20:54:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10556554872d400aebce11217e4b14e313c6ba97","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-08-25 21:08:41.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"5a943635c3ac083e3d4bc6ae9c70a734a5e28dda","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-08-25 21:26:40.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4fb1f0e84025daa33bdc733f72ea75bd64504ae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 23:21:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec10b7e39f9f4c64958f659de1653b81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a0fa698ac6248a58167bd07078f27c8 : SUCCESS in 5m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c1c32d1aae64ae2a5e850783e149c9f : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1100bc79c46401c8d6b3ce507ff9f3a : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4a2ef42ac1042f682fec68c84722a14 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/088583efa4b34afb8f9aa389c5815496 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3cc37f5a85249d9bb7a8b332e25b9c0 : SUCCESS in 3m 39s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e343324c32ff4e16bb4ab47c1b07427e : SUCCESS in 51m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6225bc1d05f14a04956b774be375c44b : FAILURE in 1h 31m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a9683cce035e44d49aa53f77a0e65f4c : FAILURE in 1h 53m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5aceeb82b689499698baaa5f75ac48c6 : SUCCESS in 33m 41s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/3ee770b69b234e07b51070b8fc858a67 : FAILURE in 21m 32s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/1579446dc0254682bc973b5cc9a992e1 : SUCCESS in 3m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/521dd6c9125b4474a957f9c736bda943 : SUCCESS in 1h 18m 57s","accounts_in_message":[],"_revision_number":3},{"id":"7d0e06cd895c8261ebbc9d48639339836aede479","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-08-25 23:24:20.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"47f1efbbf064c04b5eb572b4c573adf00f65c3ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 01:25:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/885e7a2fd340405ca11cab019eb8acae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0ffd1f6f69a4f029e825add70822108 : SUCCESS in 8m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67eada9a66674ad6b25bbe0568f67850 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bf19757ee5c4d2a93427a531f905fa3 : SUCCESS in 7m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a7397dbda724c6980762f09303e1e0e : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/24671235b8ce4a09b4bf703b71af6760 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f3e884426944b2481cd6ab6c8c21c23 : SUCCESS in 5m 31s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/146863ddf84343d09ee73c81fede7201 : SUCCESS in 1h 46m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fa8301cba397432b895ca01dc0d15097 : SUCCESS in 1h 18m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bedaf2e019364739920e6238685bbf95 : SUCCESS in 2h 00m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8b290c24266a4dbc9c556f2c15345653 : SUCCESS in 31m 47s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/d924278f93124af1a3377ae19d72dcd1 : SUCCESS in 16m 32s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f962096cfd3c4a448fa636be85327810 : SUCCESS in 5m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3b8d260afb21489ea51c1da9bfc068cc : SUCCESS in 1h 45m 12s","accounts_in_message":[],"_revision_number":4},{"id":"4cd2982f5fa97fe248166c80ff0e4c7bc2a3d7b8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-08-26 07:41:35.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"a7defda2451259424d70a7504b0c0feeeb3e865f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-08-26 08:02:15.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"132c3bdadcd5ef4eda7645e637f172abe6a54453","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-08-26 08:19:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"49ab554aade565c5557b468ac96c0d48782eff4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 08:19:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"34c3a9b418cabd3ffa9f586a3df9416a576fe6a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 10:29:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e62d39ad3a4408299c243f2d5e63915\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6484911c25af443d8332c0f2b281d344 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5860e5e7f2ef4cc28181a45f41fd72e4 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6872749aeef4e25b79fa22e7aa2a337 : SUCCESS in 2m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3caff9ff946b4901af26926f32a0d39a : SUCCESS in 3m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e3c433d560a49709dde483e53bed7ab : SUCCESS in 1h 41m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9c1035a8f18b49489bfe6ce25eda72ea : SUCCESS in 1h 34m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/52d14a2d8f4b48c097f24b4c867cf007 : SUCCESS in 2h 09m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f5a0b63f09dc4d1088ed487991d03b9e : SUCCESS in 1h 19m 57s","accounts_in_message":[],"_revision_number":4},{"id":"16a262d968e905079e6f3dcbed62942845743620","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 10:29:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b2630bb9399a129edec3b8d7c69aba314b5fc142","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 10:32:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a094eb9d0df144b383e3037d7424b26e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e7f7c4bd774e8a91b08b62a70d7c24 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"292534cca2ecb2abf4de0c5216626e4363703abc","revisions":{"3b6c33342e9bf7bf9816d58d9644fa8d2fc29f15":{"kind":"REWORK","_number":1,"created":"2025-08-25 20:54:25.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/90/958490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/90/958490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/958490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/958490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/958490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/90/958490/1"}}},"commit":{"parents":[{"commit":"6dbe4b2e8cceb8c1af52a72696f7f7eb2efc9b5d","subject":"Merge \"tox: Remove basepython and ineffective ignore_basepython_conflict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6dbe4b2e8cceb8c1af52a72696f7f7eb2efc9b5d"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-08-25 20:54:25.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-08-25 20:54:25.000000000","tz":0},"subject":"Replace \"tenant\" with project","message":"Replace \"tenant\" with project\n\nChange-Id: I7fecb3f34b4b3046587c1a1bc28f7dc7736907eb\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3b6c33342e9bf7bf9816d58d9644fa8d2fc29f15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3b6c33342e9bf7bf9816d58d9644fa8d2fc29f15"}]},"branch":"refs/heads/master"},"495e07153ca8427c9009d2e7a5e20385aba43576":{"kind":"REWORK","_number":2,"created":"2025-08-25 21:08:41.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/90/958490/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/90/958490/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/958490/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/958490/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/958490/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/90/958490/2"}}},"commit":{"parents":[{"commit":"6dbe4b2e8cceb8c1af52a72696f7f7eb2efc9b5d","subject":"Merge \"tox: Remove basepython and ineffective ignore_basepython_conflict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6dbe4b2e8cceb8c1af52a72696f7f7eb2efc9b5d"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-08-25 20:54:25.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-08-25 21:08:27.000000000","tz":0},"subject":"Replace \"tenant\" with \"project\"","message":"Replace \"tenant\" with \"project\"\n\nUpdated all warning and error messages to use the preferred term\n\"project\" instead of \"tenant\" for consistency.\n\nChange-Id: I7fecb3f34b4b3046587c1a1bc28f7dc7736907eb\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/495e07153ca8427c9009d2e7a5e20385aba43576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/495e07153ca8427c9009d2e7a5e20385aba43576"}]},"branch":"refs/heads/master"},"590109a3d3f1fe124142e1b2e86ffe9e717daad0":{"kind":"REWORK","_number":3,"created":"2025-08-25 21:26:40.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/90/958490/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/90/958490/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/958490/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/958490/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/958490/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/90/958490/3"}}},"commit":{"parents":[{"commit":"6dbe4b2e8cceb8c1af52a72696f7f7eb2efc9b5d","subject":"Merge \"tox: Remove basepython and ineffective ignore_basepython_conflict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6dbe4b2e8cceb8c1af52a72696f7f7eb2efc9b5d"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-08-25 20:54:25.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-08-25 21:26:01.000000000","tz":0},"subject":"Replace \"tenant\" with \"project\"","message":"Replace \"tenant\" with \"project\"\n\nUpdated all warning and error messages to use the preferred term\n\"project\" instead of \"tenant\" for consistency.\n\nChange-Id: I7fecb3f34b4b3046587c1a1bc28f7dc7736907eb\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/590109a3d3f1fe124142e1b2e86ffe9e717daad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/590109a3d3f1fe124142e1b2e86ffe9e717daad0"}]},"branch":"refs/heads/master"},"292534cca2ecb2abf4de0c5216626e4363703abc":{"kind":"REWORK","_number":4,"created":"2025-08-25 23:24:20.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/90/958490/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/90/958490/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/958490/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/958490/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/958490/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/90/958490/4"}}},"commit":{"parents":[{"commit":"6dbe4b2e8cceb8c1af52a72696f7f7eb2efc9b5d","subject":"Merge \"tox: Remove basepython and ineffective ignore_basepython_conflict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6dbe4b2e8cceb8c1af52a72696f7f7eb2efc9b5d"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-08-25 20:54:25.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-08-25 23:24:05.000000000","tz":0},"subject":"Replace \"tenant\" with \"project\"","message":"Replace \"tenant\" with \"project\"\n\nUpdated all warning and error messages to use the preferred term\n\"project\" instead of \"tenant\" for consistency.\n\nChange-Id: I7fecb3f34b4b3046587c1a1bc28f7dc7736907eb\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/292534cca2ecb2abf4de0c5216626e4363703abc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/292534cca2ecb2abf4de0c5216626e4363703abc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
