)]}'
{"id":"openstack%2Fneutron-lib~961612","triplet_id":"openstack%2Fneutron-lib~master~I0ab856f56ff4e03cfff30075de938ae7178ebbff","project":"openstack/neutron-lib","branch":"master","topic":"bug/2124252","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-10-02 09:17:10.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-10-02 06:57:35.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-10-02 09:17:10.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-10-01 16:14:50.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I0ab856f56ff4e03cfff30075de938ae7178ebbff","subject":"Add extension \"router-enable-snat\"","status":"MERGED","created":"2025-09-18 09:47:44.000000000","updated":"2025-10-02 09:18:03.000000000","submitted":"2025-10-02 09:17:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961612-bug/2124252","meta_rev_id":"14f9a7c24d574439a9b9c306e5c59740e92ecb8b","_number":961612,"virtual_id_number":961612,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-02 09:17:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-10-02 06:57:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-10-01 16:14:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-10-02 06:57:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-18 12:09:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-19 14:42:10.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-09-22 11:12:21.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-09-22 12:08:10.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-10-01 16:14:50.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"9a8c9bddea0a8020ad93b3875443ca28bf5d80f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-18 09:47:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a73ce38983214b365043de4ff5b7880810ff9bcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-18 10:10:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"506dac487dba92ef183e3d5a2483174f5ce9f5f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 12:09:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a1ba01a548a4c318d2ecfdf135a9ff5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/25650344d0bf493f8923ef2a96cb29fb : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e6a0155704410683d627cb5a1ae406 : SUCCESS in 6m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e60ba6d444804df8b468a3f15f6cfb19 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/877951bb8bfe447eb86e376347ef2228 : SUCCESS in 7m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/028ff6cb683442ddba49102ddf8f8e6e : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2bb5159b3fb94d4d8c8dbf312f30be22 : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dc59987ca1e3439e920dba9bac54546a : SUCCESS in 5m 58s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e165a4d306784e86a5979de6824167cc : SUCCESS in 1h 48m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5cfb3aae792941da89081a26edfd3efd : SUCCESS in 1h 32m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3ce5abef6b3741e0b4123c0e9afaa5cd : SUCCESS in 1h 45m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f1ca8bef47a74fa18f7b7a0b2b5ac098 : SUCCESS in 34m 46s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/85b49532d3214c0ba7642b6e08d265ed : SUCCESS in 25m 00s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/43ba01b88a4745789cd1114794d6dbe2 : SUCCESS in 5m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/82647179cd9c4b608b7c03b8c00c335e : SUCCESS in 1h 34m 37s","accounts_in_message":[],"_revision_number":2},{"id":"068e2ef467bbc3052e76fff1acf9de2fe5b4aaa3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-09-19 14:42:10.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cf8a4aed4174c46ea1eb1df7401f36076e54c998","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-22 08:19:32.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d20943486c72d4ec3ab6a81c53a0d93deb7fb8c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-22 08:20:47.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e7e4116174755cb727086bfddc5ea94a0912f56d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 10:14:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/173ab5dc80684625abe356c12d94b844\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6111ed7e990741ffa4add5213b70b8e1 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfe829bd44274c42a766462b04788b5e : SUCCESS in 9m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1623597e1d849e5b998c19525c39819 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee28ae5dc85d44429bb229166b588e97 : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b21ae19e16dd4d86b1ab8ba1a49e2e90 : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/335c7bde488644a88170c3dfd646b612 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e1a8bcb1ee94953a970f0a53b994313 : SUCCESS in 5m 22s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2650c6332b934b65abd9b7c5fbf61c10 : SUCCESS in 1h 28m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dead0f8fc95642179d4aa5c29074b637 : SUCCESS in 1h 28m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0f4e354f0e954f12a40b56a25e881e6f : SUCCESS in 1h 52m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/905cf859dffc4f3e9028757e6cba9334 : SUCCESS in 31m 32s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/baecaf0be0d043e3927f9e084bd67f97 : SUCCESS in 20m 02s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9b8cc608a2b44969b516fb527e13a4b1 : SUCCESS in 5m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e17584f5cf95466689ed75150523da17 : SUCCESS in 1h 28m 03s","accounts_in_message":[],"_revision_number":3},{"id":"ee77dc10b01bb47030545356c47b65e73b461f08","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-22 11:12:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a7ed1a52af79a26b50f71d5a4a6f4764f4df2878","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-22 12:08:10.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d2d338408713e31b25ec8907c774913ca36ca2f9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-22 14:41:45.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6dd5195420013653468dead892f95f0de43b1f9c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-22 14:41:48.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"63d4e841b4ee664d11c5fa65a175432569d0cb0f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-22 14:49:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"784dc52f697ba5107ba5163c6edcc5d0d62cb84a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-22 14:50:45.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"356316e5cad909ba806b35c055af8a7b09818b2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 16:36:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3435386427f1456881e431a846f58788\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9d112a1294cc48c291d1a58efdc5f763 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19d02178bb7348ecbd2ded973aa354b1 : SUCCESS in 9m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e16592aed6a341428bd187d8d72f69f3 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b95ce2489b584a758d3f80e30ed44ac3 : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dba6fe6b61c844d6acc2e6dfd38ba05b : SUCCESS in 6m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/161523d1efa2430ca36337d2ff7f0014 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c131c1f53cc64f64aa73e6be4be1781b : SUCCESS in 6m 29s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e848b24334ba4718a87527b9147260d8 : SUCCESS in 1h 33m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/019c7e7cbf594e1a991a50d4842856d9 : SUCCESS in 1h 38m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5863b1541c394916bcf8b407262b5825 : SUCCESS in 1h 44m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/de35e7479f4743e3b28d0b8051ae5ac6 : SUCCESS in 33m 25s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/d74b061d782e4832bba292aea3aad237 : SUCCESS in 20m 06s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ba3352728bc946bfb1b2d2b524dc8308 : SUCCESS in 5m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f38aeb41876246aabd7fe5de9a7c30a2 : SUCCESS in 1h 31m 45s","accounts_in_message":[],"_revision_number":4},{"id":"60284b0b7d186bd588dd6e9ea129348bbc1a2df9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-01 16:14:50.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1dc6abe30618318f2c92b156cca3e0b8d119b0fa","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-10-02 06:57:35.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9c6d41d1e8b8632d54c526b6bc208b9d1ed8a5e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 06:57:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"816f1d1615e4eac99a970a2d7f134144179798be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 09:17:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b02f924d053942878e85cc55c038a288\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/49e56c6de5e042db90f539d940a36c5e : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79e9a8bdb80b4575857484b12ef394b4 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da5ef2d85a044725ab844613ad78d144 : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7bc3efa4ac86490188edb0e9cd91e4b3 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a181112b264b4b8aaede03e84e69a27f : SUCCESS in 3m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0e2c915020547ce9806c4ffa9784f3b : SUCCESS in 2h 03m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6b2b6913e6b145b2ade5e16c348ed6c4 : SUCCESS in 1h 16m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6d5a67f8ec664c8dacaf655b54cd1573 : SUCCESS in 2h 18m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/423377abf65e41fcb0b7e037135578e9 : SUCCESS in 2h 06m 19s","accounts_in_message":[],"_revision_number":4},{"id":"41eb444f46104f01fd17c40d4154bacd8a96321f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 09:17:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"14f9a7c24d574439a9b9c306e5c59740e92ecb8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 09:18:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4935f501a69470f8c69845010094c56\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/96e7f16070734fb2809b4b79226f6942 : SUCCESS in 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98d2fed2620a421ea5ec88fcdb6b4412 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7e78006e5ad78dfd14c718d32fc6c74618ee2dac","revisions":{"8c568d231fff3753418186c7350b27d5f0db513d":{"kind":"REWORK","_number":1,"created":"2025-09-18 09:47:44.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/12/961612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/12/961612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/961612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/961612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/961612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/12/961612/1"}}},"commit":{"parents":[{"commit":"718b9c07aa409b02ef1a2c879a8f85a0c1697e4b","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/718b9c07aa409b02ef1a2c879a8f85a0c1697e4b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-18 09:45:55.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-18 09:45:55.000000000","tz":0},"subject":"Add extension \"router-enable-snat\"","message":"Add extension \"router-enable-snat\"\n\nThis new extension exposes the \"enable_snat\" flag of a router.\n\nRelated-Bug: #2124252\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I0ab856f56ff4e03cfff30075de938ae7178ebbff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8c568d231fff3753418186c7350b27d5f0db513d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8c568d231fff3753418186c7350b27d5f0db513d"}]},"branch":"refs/heads/master"},"f42aff9ce83ac815100dc42837be5558bf49ac88":{"kind":"REWORK","_number":2,"created":"2025-09-18 10:10:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/12/961612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/12/961612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/961612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/961612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/961612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/12/961612/2"}}},"commit":{"parents":[{"commit":"718b9c07aa409b02ef1a2c879a8f85a0c1697e4b","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/718b9c07aa409b02ef1a2c879a8f85a0c1697e4b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-18 09:45:55.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-18 10:08:25.000000000","tz":0},"subject":"Add extension \"router-enable-snat\"","message":"Add extension \"router-enable-snat\"\n\nThis new extension exposes the \"enable_snat\" flag of a router. The\n\"enable_snat\" flag can be updated using the external gateway API\ncommands. However this is a router global configuration property\nthat is stored directly on the router register. As described in the\n``update-external-gateways-of-router`` API section [1]:\n\n  \"The enable_snat field does not have any effect for extra gateways\n  except for the first external gateway in the list.\"\n\n[1]https://docs.openstack.org/api-ref/network/v2/index.html#update-external-gateways-of-router\n\nRelated-Bug: #2124252\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I0ab856f56ff4e03cfff30075de938ae7178ebbff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f42aff9ce83ac815100dc42837be5558bf49ac88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f42aff9ce83ac815100dc42837be5558bf49ac88"}]},"branch":"refs/heads/master"},"d98edd3e05a44e071ddb78b6fd54d35f7a64d7c3":{"kind":"REWORK","_number":3,"created":"2025-09-22 08:20:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/12/961612/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/12/961612/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/961612/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/961612/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/961612/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/12/961612/3"}}},"commit":{"parents":[{"commit":"718b9c07aa409b02ef1a2c879a8f85a0c1697e4b","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/718b9c07aa409b02ef1a2c879a8f85a0c1697e4b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-18 09:45:55.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-09-22 08:20:47.000000000","tz":0},"subject":"Add extension \"router-enable-snat\"","message":"Add extension \"router-enable-snat\"\n\nThis new extension exposes the \"enable_snat\" flag of a router. The\n\"enable_snat\" flag can be updated using the external gateway API\ncommands. However this is a router global configuration property\nthat is stored directly on the router register. As described in the\n``update-external-gateways-of-router`` API section [1]:\n\n  \"The enable_snat field does not have any effect for extra gateways\n  except for the first external gateway in the list.\"\n\n[1]https://docs.openstack.org/api-ref/network/v2/index.html#update-external-gateways-of-router\n\nRelated-Bug: #2124252\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I0ab856f56ff4e03cfff30075de938ae7178ebbff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d98edd3e05a44e071ddb78b6fd54d35f7a64d7c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d98edd3e05a44e071ddb78b6fd54d35f7a64d7c3"}]},"branch":"refs/heads/master"},"7e78006e5ad78dfd14c718d32fc6c74618ee2dac":{"kind":"REWORK","_number":4,"created":"2025-09-22 14:50:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/12/961612/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/12/961612/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/961612/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/961612/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/961612/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/12/961612/4"}}},"commit":{"parents":[{"commit":"718b9c07aa409b02ef1a2c879a8f85a0c1697e4b","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/718b9c07aa409b02ef1a2c879a8f85a0c1697e4b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-18 09:45:55.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-22 14:50:20.000000000","tz":0},"subject":"Add extension \"router-enable-snat\"","message":"Add extension \"router-enable-snat\"\n\nThis new extension exposes the \"enable_snat\" flag of a router. The\n\"enable_snat\" flag can be updated using the external gateway API\ncommands. However this is a router global configuration property\nthat is stored directly on the router register. As described in the\n``update-external-gateways-of-router`` API section [1]:\n\n  \"The enable_snat field does not have any effect for extra gateways\n  except for the first external gateway in the list.\"\n\n[1]https://docs.openstack.org/api-ref/network/v2/index.html#update-external-gateways-of-router\n\nRelated-Bug: #2124252\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I0ab856f56ff4e03cfff30075de938ae7178ebbff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e78006e5ad78dfd14c718d32fc6c74618ee2dac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e78006e5ad78dfd14c718d32fc6c74618ee2dac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
