)]}'
{"id":"openstack%2Fneutron-lib~962906","triplet_id":"openstack%2Fneutron-lib~master~Ie633c551936ffb9c4a4b294a5b52cec4accdfb09","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"31746":{"account":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"last_update":"2025-12-04 13:55:25.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie633c551936ffb9c4a4b294a5b52cec4accdfb09","subject":"add reason field for PortInUse exception","status":"ABANDONED","created":"2025-10-03 07:58:05.000000000","updated":"2025-12-04 13:55:25.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"396d9c5c4631810e7a755ece0d517847b22a6be8","_number":962906,"virtual_id_number":962906,"owner":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-10-03 10:12:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":-1,"date":"2025-10-06 07:14:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-03 10:12:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-03 10:48:03.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-10-03 14:35:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-10-06 07:14:03.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"581faa5b04e113acf5d003c9d799d163fe1ee731","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2025-10-03 07:58:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"354e1b7d4eb99bffb4cbba12b7a4d41b4c824c71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2025-10-03 08:07:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fc30622e4dab47384a179488a33db93420b4817b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 10:12:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fde1001c2e44ad0997358d9c304db79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ccb9c32d83043e7a46e3de598d5f4ff : SUCCESS in 5m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a219720ef9834104a9ee1d67c60b9180 : FAILURE in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f047a5e3d2349458aba59a27cc0a012 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68027b64ecad408ba702132469017217 : FAILURE in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70c0c5f6e87b48888fddecbbba6fdc21 : FAILURE in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a051dfbde704821b1f24134d6b5efb7 : FAILURE in 5m 37s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea0f84e5aac843ea956ccde91d732326 : SUCCESS in 1h 59m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c3c5f126d594052b49cc94d963a15d7 : SUCCESS in 3m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7e3b398ca85b49fb8877cd7804dc123e : SUCCESS in 1h 32m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/60da49aeadcf418db04fee4d69ef91d9 : SUCCESS in 1h 54m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/485136a2337b44ee86c6dc8bc75340c1 : SUCCESS in 35m 13s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/066e7a0de96f4abba756bb7933f2ea5f : SUCCESS in 24m 21s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/02256f37bfa149d58dffcdb9ff2629a7 : FAILURE in 5m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d2826bd4c0134954b32ab4eea0d4d628 : SUCCESS in 1h 51m 08s","accounts_in_message":[],"_revision_number":2},{"id":"623bb23815c8452fb049254e1e53ca29839f8720","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-10-03 10:48:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"48d58ec5ca1aaa43dca0314b3dceae397a52a8cc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-10-06 07:14:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"396d9c5c4631810e7a755ece0d517847b22a6be8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-04 13:55:25.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"01fbe483eedbfd0bb3eadd71ec4a7c49149473fb","revisions":{"03621ce1eabcb8e68c5874837a72967189fcfb0f":{"kind":"REWORK","_number":1,"created":"2025-10-03 07:58:05.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/06/962906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/06/962906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/06/962906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/06/962906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/06/962906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/06/962906/1"}}},"commit":{"parents":[{"commit":"7e78006e5ad78dfd14c718d32fc6c74618ee2dac","subject":"Add extension \"router-enable-snat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e78006e5ad78dfd14c718d32fc6c74618ee2dac"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2025-10-03 07:53:52.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2025-10-03 07:57:55.000000000","tz":180},"subject":"add reason field for PortInUse exception","message":"add reason field for PortInUse exception\n\nChange-Id: Ie633c551936ffb9c4a4b294a5b52cec4accdfb09\nSigned-off-by: okozachenko \u003cokozachenko@vexxhost.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/03621ce1eabcb8e68c5874837a72967189fcfb0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/03621ce1eabcb8e68c5874837a72967189fcfb0f"}]},"branch":"refs/heads/master"},"01fbe483eedbfd0bb3eadd71ec4a7c49149473fb":{"kind":"REWORK","_number":2,"created":"2025-10-03 08:07:35.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/06/962906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/06/962906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/06/962906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/06/962906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/06/962906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/06/962906/2"}}},"commit":{"parents":[{"commit":"7e78006e5ad78dfd14c718d32fc6c74618ee2dac","subject":"Add extension \"router-enable-snat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e78006e5ad78dfd14c718d32fc6c74618ee2dac"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2025-10-03 07:53:52.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2025-10-03 08:07:25.000000000","tz":180},"subject":"add reason field for PortInUse exception","message":"add reason field for PortInUse exception\n\nChange-Id: Ie633c551936ffb9c4a4b294a5b52cec4accdfb09\nSigned-off-by: okozachenko \u003cokozachenko@vexxhost.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/01fbe483eedbfd0bb3eadd71ec4a7c49149473fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/01fbe483eedbfd0bb3eadd71ec4a7c49149473fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
