)]}'
{"id":"openstack%2Fneutron-lib~965087","triplet_id":"openstack%2Fneutron-lib~master~Ia6e264c8318636f22ae26c60589054944df66542","project":"openstack/neutron-lib","branch":"master","topic":"bug/2107316","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-12-09 14:29:33.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"36760":{"account":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"last_update":"2025-12-10 12:22:24.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-11-10 09:53:39.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-12-04 15:29:59.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ia6e264c8318636f22ae26c60589054944df66542","subject":"Add information to network ip availabilities","status":"MERGED","created":"2025-10-28 11:06:32.000000000","updated":"2025-12-10 12:23:20.000000000","submitted":"2025-12-10 12:22:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"965087-bug/2107316","meta_rev_id":"9e635316ce6e712d6c2105042bdd0a01df469f5b","_number":965087,"virtual_id_number":965087,"owner":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-10 12:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-12-09 14:29:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-12-10 10:29:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-12-05 15:13:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-12-10 10:29:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-28 12:56:13.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-10-28 13:27:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-28 18:43:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-12-03 07:54:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-12-03 22:29:21.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"9445b361532fc7da386837406b1db863748acf7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2025-10-28 11:06:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"517c6c0940884e7b8680e4997d3bee5f6ef91999","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-10-28 12:56:13.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6af73c25c3272ccc7bc248502594c69bc90ab13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 13:27:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/581bc4dd0f614cd1acb2ffcbd6105f20\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7babc50aa97a49d7b9af78a4881de8ec : FAILURE in 2m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8a527d6794e4ad9b736c509bafb17c4 : SUCCESS in 3m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7f12bb8d52d48cdb0fa27ad30060b4f : FAILURE in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d01044b0a022419a988b3bea9278443d : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/362f85d1087d445fa0fc5a59a793079d : FAILURE in 2m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70d007832fcb4ec9856b246cd89981c3 : FAILURE in 5m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d0315ecf53b448ebfb3c1cc9a63d275 : SUCCESS in 1h 51m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7b8df0da619749aab0b279bb8c91daad : SUCCESS in 1h 29m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4157957026fc44ad8c6d84f3039ce88b : SUCCESS in 2h 20m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/be67c54d5e82465c9724aa7bc3bc0ab6 : SUCCESS in 37m 10s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/3cab7dbc58174d129b365f59b827a02e : SUCCESS in 18m 54s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d2ef8b231b4b4667b3841a8ed5f849f0 : FAILURE in 5m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d6d1e12f2bf54939910154c32a85cabc : FAILURE in 32m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3cf0eb742c46eabc0d67dfc7f8af8854116647f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2025-10-31 09:30:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2dda59db011c2d61eb875381a65acd3b8934a306","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2025-10-31 09:33:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ea6d1473ae3661a47b7914c1460f230b63610e24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 11:21:54.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/252793bdca6e489abff8d2e7fb6aa909\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d1c6b0ce9dd8436f8e6f13d36e4627f1 : FAILURE in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d7d8eee097b42f3b68b2996d5128afe : SUCCESS in 9m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/488aad641ca14182aa229911906f5d7e : FAILURE in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c77ba436699e47f099ba5e8078f53235 : FAILURE in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/edcc8cbfc22441c7b0f4eac0200bc4a6 : FAILURE in 2m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11d0258d7a7f42ae820ada13681e5e69 : FAILURE in 6m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3314ccd493649a1abcdb8574df85932 : SUCCESS in 52m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1374080c312a4cbbbae6cc47c4f91352 : SUCCESS in 1h 26m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/22347e68e2654c71ae53278e1e7a847b : SUCCESS in 1h 50m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/59c49bbe8e404072837f89b9f118acd6 : SUCCESS in 34m 01s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/b7951584bfd346d7aa2161d32de159e0 : SUCCESS in 25m 41s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0bd829475f2b4914933ed6a70008da69 : FAILURE in 6m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9fec666c18614feb9fb9830eabcb0810 : SUCCESS in 1h 31m 46s","accounts_in_message":[],"_revision_number":2},{"id":"dda153756f0512afd19f3c47ffc78fb50d89037a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2025-11-03 06:29:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"35f6f9208d6c40df786c8769ff67727003c98e5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 09:20:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac71a3e37eb349b1a8fb5980f7f6eb7e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bbec6fd2af2b45a0b007d51d0d8e9596 : FAILURE in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c712c1cedd4250a5dd4ec5f5e81dcf : SUCCESS in 8m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/065f001c6d4c4db285d2e09b27c1e6fa : FAILURE in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2c562fbda5b4190a35e510283c02baf : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1bf0481da8824fd09ac4a3f548a92aca : FAILURE in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad9ac2dad48f4713b041b2954f7dc61f : FAILURE in 6m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d261f64b709c4329a354c23293edadb2 : SUCCESS in 1h 19m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2631c17bece94598861c810de2e4f267 : TIMED_OUT in 2h 49m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e16b43cf0d2c4255aee2a8fc7ceb979e : SUCCESS in 1h 48m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4d8080bef70b409eb73d6f3a16eadeb6 : SUCCESS in 31m 25s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/890a86d0b4544b84a8b1392262fcc737 : SUCCESS in 36m 08s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e0cdcc7f72c74a3d8931f53edf2d8a18 : FAILURE in 4m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/85257448145242cead553225bb58a810 : SUCCESS in 2h 10m 43s","accounts_in_message":[],"_revision_number":3},{"id":"dad72e8280f07f2a4e6f7d741a0e3ac878bac8f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2025-11-04 12:41:09.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c1daaaded28417b7f395c18f031c0c23bf6804a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 14:58:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/11ced311d1ad418fb2819089181fc507\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/03b0b4aff8544b5cbc6fe8bb047acf1a : FAILURE in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dbcd501507240a6ba298551e0fa3f6a : SUCCESS in 8m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfd2d3be40b14a05b2a43f47d492c9a5 : FAILURE in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d98642b84424c8d814283294730c030 : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2130d920e1fe4b5fbab20fd545ac093a : FAILURE in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01e7c65247f144789a84ba3480974c05 : FAILURE in 5m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/678243d16dee420f8aee17ddf880595c : SUCCESS in 1h 33m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6960667469744cafaec876e4b855b254 : SUCCESS in 1h 30m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2377a33671aa4d76a7942e7f39b4d13c : SUCCESS in 2h 15m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/15c929e1701a4aeeb978a969aa518d16 : SUCCESS in 34m 38s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/292dd7c51bf14a5fa235d057734a88f6 : SUCCESS in 19m 54s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d464c95629524fe1a932a1861dec9bff : FAILURE in 5m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/aca1254d03104494b5313a42504cd604 : SUCCESS in 1h 41m 43s","accounts_in_message":[],"_revision_number":4},{"id":"cc548536c3416763d596e7bb244c616047e59c3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2025-11-05 12:12:06.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"c723f2ebe6023da47eb165c440ab6e3254caf3e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 14:31:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8dc2fd07ba548048044c11b7d148a9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/562f8917afba40b8a016981d24b86764 : SUCCESS in 6m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7b59a7783234417b09b0dd342d12062 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1710f39d1c7459a92378f103f21b2e3 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1b8091609ed4c54b79d5791334a549c : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83703e7b8e2a458092ec1cddf103f07a : SUCCESS in 6m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/435ca8e4898e4021884c1496c205d8f2 : SUCCESS in 1h 47m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3783912fe4f843ef9d1c3604781498ef : SUCCESS in 1h 25m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e6843d98bfeb4085a7b7845851726948 : SUCCESS in 2h 18m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a47f64e438a4430f899f7ef27f0f53fc : SUCCESS in 30m 49s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/55643993b3d1407c802886ccbae01c42 : SUCCESS in 20m 10s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/91ca2709e11b486db1504368632fbf5b : SUCCESS in 3m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2e432d1714c44c42a93258d7fca03e6a : SUCCESS in 1h 14m 49s","accounts_in_message":[],"_revision_number":5},{"id":"77531143831af7b21b0b2fd3f63c04a303b8ada1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-10 09:53:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e8abfeaf4da7054151899d91580e4d2f5a243979","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2025-11-11 05:50:53.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"547d99dd683de3fb89c73b8ff7ddf30e29f9bc92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 07:42:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e021d1e9082646138962e92de0124fa9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f4ed33f7f41d4a4abb64fac3a654bec1 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a59cdfd913154edfa458aa4757e5bc08 : SUCCESS in 7m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10ab6a6b67a84417b717eed28eb47734 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9316e82f0209433582152732f72f9323 : SUCCESS in 7m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eae0d27538964956be8d47e3d832424e : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/026f0f8b380941de8589ace3cbe81a16 : SUCCESS in 4m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b26baecff9a54435873e31a8e641e0cb : SUCCESS in 1h 21m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dd4929a5ad5844bf97cfab0250cc0752 : SUCCESS in 1h 30m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f6918ecb7a224828a0a49b7cf31f3882 : SUCCESS in 1h 48m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0c6398551df84a7688de39b980484285 : SUCCESS in 31m 06s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/90d232b15e494e0cb6fedd928819827c : SUCCESS in 30m 02s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5855ccd618714d8e9bcc3b801be418b0 : SUCCESS in 5m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c63863296ab54fa6a8b89be06882e5b1 : SUCCESS in 1h 46m 37s","accounts_in_message":[],"_revision_number":6},{"id":"4905559fff9d19bb0fb62f406d20336e843a4e7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2025-12-03 05:50:28.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"4025504f4a3eb9be78d286f9d6e22d5896d153dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 07:49:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48aaa2da00fa4393b227002f5022784a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5c79dfa7fd704696a19dbf77fe061a7d : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8511f72715d14b0e837c7afc12d28bdb : SUCCESS in 7m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31ddad9956bb4a17a232ebd969715e63 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/719f693ff8cd4acc84d159112a13bbb1 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/162a4d370ecb44cab013e3f859d07c50 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/079357f300cb446784d3354c21e72e6f : SUCCESS in 6m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/060c43fe668d4b8b9f1b0a680a893c6e : SUCCESS in 1h 45m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4794c34b0a7f478382c1c6aea8e83081 : SUCCESS in 1h 20m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f20123020b1d49eb94d4535c565ebffd : SUCCESS in 1h 47m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/55697b25477544998529f6aac8476e80 : SUCCESS in 31m 28s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/fa7290891c9b42668b254be654b981db : SUCCESS in 27m 39s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9f53e8e813444bb9ba8ed3e34fc48a90 : SUCCESS in 4m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5207002c3b7a421fa15549119e0e9fb4 : SUCCESS in 1h 48m 16s","accounts_in_message":[],"_revision_number":7},{"id":"d0c25044dfcd68e99aaef3e7c85134f0dfd639a8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-03 07:54:55.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a6509782bcd1ce35ac6e39160cbdcd8e23fd4144","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2025-12-03 09:59:59.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"6a6550891807906736f6b0922ce503d54024054a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-03 11:13:05.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"6a259c6b3a5b76caea5089099cb24b3429a387c9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-03 11:13:08.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"7f7b956b612fb900990e23eb3b7afd7d38276e81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 13:21:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1490c445da34738876c4986c98cdee1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8c6287b3c51a44bbb884f2edc3b1350b : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d4b46981474542ba9966451b95e67b : SUCCESS in 9m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee7c7245766244e6b64b62ba5fe3c4f2 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e94d07b003a847978b4435154f437f0f : SUCCESS in 6m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bbdf2489424f42baae4a2d2277e34e2a : SUCCESS in 2m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50b7610ef39f4872ae96fa30c6f24869 : SUCCESS in 6m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80a84718b94d46fbbb9693ad4503e31c : SUCCESS in 1h 46m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b84b4d6bafd145168250358d95da4708 : SUCCESS in 3m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a01b19b57d084872b9ac95581e3c631a : SUCCESS in 1h 18m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a630f0e1c42745c7bcd595e6218a26f7 : SUCCESS in 2h 07m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0e59d48c5b2f4c6cb45d38d6453a0728 : SUCCESS in 31m 41s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/ac20b388eaf34be8811ad468ddafa3cf : SUCCESS in 21m 07s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a84fd30614954049af5b1e575242444b : SUCCESS in 5m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/aadaab9bcf244221ad8232358aaa3e2e : SUCCESS in 1h 53m 28s","accounts_in_message":[],"_revision_number":9},{"id":"6533358787cb49373be05094a07563128e8fa239","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-03 22:29:21.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5289f66ce10b16da0ad92cf5f128a9433e0be1c1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-03 22:38:46.000000000","message":"Patch Set 9:\n\n(7 comments)","accounts_in_message":[],"_revision_number":9},{"id":"551f2e4a99660a64485fe2f96c2f91dd144cb976","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2025-12-04 04:24:26.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"702a58e0d3541440420fd7812bd5bf5c283f4918","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2025-12-04 04:25:09.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"556ad5e80d01979d37e975167dd287a4602b2247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 06:39:02.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e3803f0848a4b5b8f6285c3ad67a3c0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d78c27535c5a400790c17808b7b5d794 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51fa35a427dd48678ab9807b55bca7f8 : SUCCESS in 8m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e60507a261f4e72a8960555f553b653 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8852eb322af42c69297c7f48ff98eab : SUCCESS in 7m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/193b213c41034f63a5544f356fb10350 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/33de280e34584ac3ba947d894361d396 : SUCCESS in 5m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df0cfd613d614970a0f1bee2028b8114 : SUCCESS in 2h 03m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fef7c9c820af4ca99bc807edf13a9a35 : SUCCESS in 4m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5080cb5370d94207b7e334ff98f0359f : SUCCESS in 1h 38m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a45e69b30e3c4086821be1f237fdd594 : SUCCESS in 2h 13m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/03a9b19170274fc1a746ea9fa2424ffb : SUCCESS in 30m 52s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/3af07615d8154bd78479e2964bd0a86f : SUCCESS in 20m 40s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3f40941f383542519b474efc34d17ba8 : SUCCESS in 5m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f2249f182a6c409689d396b42ae4dca8 : SUCCESS in 1h 11m 25s","accounts_in_message":[],"_revision_number":10},{"id":"9c6f7ab06dcfcdf88d3d4af22cf448cca817640e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-04 15:29:59.000000000","message":"Patch Set 10:\n\n(7 comments)","accounts_in_message":[],"_revision_number":10},{"id":"dbce2ea2e19e2e4a3295e261d3df4882de4474ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2025-12-05 06:26:30.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"4d10505668554875413d80dd76715de1d50c20e3","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2025-12-05 06:32:53.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7490144f06d41a91c3d50b6216f7907eeb712014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 08:27:18.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb6bcedc492547a18ddf8af7f42ed1c9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8a2e7bce5e234dd49b88f66436942f20 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa02ccabc7bd4931a9f9282ef287fd32 : SUCCESS in 8m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cc5fe457adc4b2388a9b7b86f2d5aea : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07cbc2491c124c798e63dc2838e60bde : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f268e7ed723a40f9a8de1dd3e9b3ad2e : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5bebac6fc2d449ffb18b39eb4ce55103 : SUCCESS in 6m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e71816657664c8bbf27c26688b885ac : SUCCESS in 1h 52m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a3921949e7a4c8ba4273f7ed92ee19a : SUCCESS in 4m 13s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a6aaffea8acb4bcbb2c9ef4f2f76bcb5 : SUCCESS in 1h 17m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ccdfe9ac710746b0899dde9c6f5c4bdb : SUCCESS in 1h 59m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2aec95f3b4ab4183b81a593e88be71f6 : SUCCESS in 31m 48s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/cb5887dd3b42465aba86afbb7fc8b8fb : SUCCESS in 28m 55s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d7feacf78cee42cc91669430ecfaaf9f : SUCCESS in 5m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8480552391c44a818f9a71184773991b : SUCCESS in 1h 48m 04s","accounts_in_message":[],"_revision_number":11},{"id":"7e1eb91695413d6111a94a60373817f35dd974f1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-05 15:13:10.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"507c07f6e99f832182c8c40e52b9bab62b483906","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-09 14:29:33.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"cfd9f3bb9abad4f143659dbeb4e74aae270b06a3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-12-10 10:29:59.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"fb236b3c7dbaef181299b134019eed6be3b84eda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 10:30:20.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"17bb8d305ea45369578dea19293796dbd57d7d1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 12:22:24.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c89fee469a248e9b65e418cba412a5d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f78feb42e9614323959f50874859cc6f : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a15e940df2f54afeb82a8b83fa095ede : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fada409c0a945feb3322e26a0ed9804 : SUCCESS in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39231295536f4fa59d34c4b8cc1fe31c : SUCCESS in 2m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/006fcd012af04ed49fcac4257d38a7cb : SUCCESS in 5m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ccffa1e117b4f998f884dfb3cf96983 : SUCCESS in 54m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c79c29caa225487f99cd92080f701e93 : SUCCESS in 3m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/594ca5483f744bdaab9cc80ff08594ae : SUCCESS in 1h 29m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/693c079bcda949b0bef63925dfc8a360 : SUCCESS in 1h 51m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/55952c7b98564c2bb963580399b7ca7e : SUCCESS in 1h 45m 23s","accounts_in_message":[],"_revision_number":11},{"id":"48a3d123e68ab4eed08b6dae6984f9608b46eec8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 12:22:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"9e635316ce6e712d6c2105042bdd0a01df469f5b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 12:23:20.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aad5d52a1f34a91ac5475150558bc54\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/63bb6f671af147fc97dfa84d573550a1 : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f1e79a7e4bf46df8038589eabb22c8f : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/605c58f3bd714153a82979c272b2d3e3 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"b7dd86ea0c6857f48010af63941766281b5a4570","revisions":{"b6257d7c8eff0640298673e31213cb89e7fa3658":{"kind":"REWORK","_number":1,"created":"2025-10-28 11:06:32.000000000","uploader":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"ref":"refs/changes/87/965087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/965087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/1"}}},"commit":{"parents":[{"commit":"7b3be70b350c3cbe4298bc05ae648fe5fadb19b4","subject":"Fix duplicate registration for resource_extend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7b3be70b350c3cbe4298bc05ae648fe5fadb19b4"}]}],"author":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-10-24 11:17:25.000000000","tz":540},"committer":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-10-28 11:03:50.000000000","tz":540},"subject":"Add information to network ip availabilities","message":"Add information to network ip availabilities\n\nThe change[0] provides full information on the number of ips in the\nsubnet from the response of GET ``/v2.0/network-ip-availabilites``\n- the number of ips in the subnet cidr, in the allocation pools, used\nips in the subnet cidr, and used ips in the allocation pools.\n\nThis change is for adding fields to the api result.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/965083\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/947826\n\nRelated-Bug: #2107316\nChange-Id: Ia6e264c8318636f22ae26c60589054944df66542\nSigned-off-by: Jimin Shin \u003cjimin3.shin@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b6257d7c8eff0640298673e31213cb89e7fa3658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b6257d7c8eff0640298673e31213cb89e7fa3658"}]},"branch":"refs/heads/master"},"eb4e0979efd545aa1ce987f1d0caaac6266c5793":{"kind":"REWORK","_number":2,"created":"2025-10-31 09:30:28.000000000","uploader":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"ref":"refs/changes/87/965087/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/965087/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/2"}}},"commit":{"parents":[{"commit":"7b3be70b350c3cbe4298bc05ae648fe5fadb19b4","subject":"Fix duplicate registration for resource_extend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7b3be70b350c3cbe4298bc05ae648fe5fadb19b4"}]}],"author":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-10-24 11:17:25.000000000","tz":540},"committer":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-10-31 09:27:23.000000000","tz":540},"subject":"Add information to network ip availabilities","message":"Add information to network ip availabilities\n\nThe change[0] provides full information on the number of ips in the\nsubnet from the response of GET ``/v2.0/network-ip-availabilites``\n- the number of ips in the subnet cidr, in the allocation pools, used\nips in the subnet cidr, and used ips in the allocation pools.\n\nThis change is for adding fields to the api result.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/965083\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/947826\n\nRelated-Bug: #2107316\nChange-Id: Ia6e264c8318636f22ae26c60589054944df66542\nSigned-off-by: Jimin Shin \u003cjimin3.shin@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/eb4e0979efd545aa1ce987f1d0caaac6266c5793"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/eb4e0979efd545aa1ce987f1d0caaac6266c5793"}]},"branch":"refs/heads/master"},"e6ef7c71cff0c9d445539d8ec36169184d4808a9":{"kind":"REWORK","_number":3,"created":"2025-11-03 06:29:49.000000000","uploader":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"ref":"refs/changes/87/965087/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/965087/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/3"}}},"commit":{"parents":[{"commit":"7b3be70b350c3cbe4298bc05ae648fe5fadb19b4","subject":"Fix duplicate registration for resource_extend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7b3be70b350c3cbe4298bc05ae648fe5fadb19b4"}]}],"author":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-10-24 11:17:25.000000000","tz":540},"committer":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-11-03 06:24:29.000000000","tz":540},"subject":"Add information to network ip availabilities","message":"Add information to network ip availabilities\n\nThe change[0] provides full information on the number of ips in the\nsubnet from the response of GET ``/v2.0/network-ip-availabilites``\n- the number of ips in the subnet cidr, in the allocation pools, used\nips in the subnet cidr, and used ips in the allocation pools.\n\nThis change is for adding fields to the api result.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/965083\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/947826\n\nRelated-Bug: #2107316\nChange-Id: Ia6e264c8318636f22ae26c60589054944df66542\nSigned-off-by: Jimin Shin \u003cjimin3.shin@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e6ef7c71cff0c9d445539d8ec36169184d4808a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e6ef7c71cff0c9d445539d8ec36169184d4808a9"}]},"branch":"refs/heads/master"},"da85c267254dca76f363a50df29b12499fb097d4":{"kind":"REWORK","_number":4,"created":"2025-11-04 12:41:09.000000000","uploader":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"ref":"refs/changes/87/965087/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/965087/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/4"}}},"commit":{"parents":[{"commit":"7b3be70b350c3cbe4298bc05ae648fe5fadb19b4","subject":"Fix duplicate registration for resource_extend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7b3be70b350c3cbe4298bc05ae648fe5fadb19b4"}]}],"author":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-10-24 11:17:25.000000000","tz":540},"committer":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-11-04 12:39:13.000000000","tz":540},"subject":"Add information to network ip availabilities","message":"Add information to network ip availabilities\n\nThe change[0] provides full information on the number of ips in the\nsubnet from the response of GET ``/v2.0/network-ip-availabilites``\n- the number of ips in the subnet cidr, in the allocation pools, used\nips in the subnet cidr, and used ips in the allocation pools.\n\nThis change is for adding fields to the api result.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/965083\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/947826\n\nRelated-Bug: #2107316\nChange-Id: Ia6e264c8318636f22ae26c60589054944df66542\nSigned-off-by: Jimin Shin \u003cjimin3.shin@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/da85c267254dca76f363a50df29b12499fb097d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/da85c267254dca76f363a50df29b12499fb097d4"}]},"branch":"refs/heads/master"},"68bfa5b1faa19e75776fdb22f9323dc5108018ed":{"kind":"REWORK","_number":5,"created":"2025-11-05 12:12:06.000000000","uploader":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"ref":"refs/changes/87/965087/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/965087/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/5"}}},"commit":{"parents":[{"commit":"7b3be70b350c3cbe4298bc05ae648fe5fadb19b4","subject":"Fix duplicate registration for resource_extend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7b3be70b350c3cbe4298bc05ae648fe5fadb19b4"}]}],"author":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-10-24 11:17:25.000000000","tz":540},"committer":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-11-05 12:11:56.000000000","tz":540},"subject":"Add information to network ip availabilities","message":"Add information to network ip availabilities\n\nThe change[0] provides full information on the number of ips in the\nsubnet from the response of GET ``/v2.0/network-ip-availabilites``\n- the number of ips in the subnet cidr, in the allocation pools, used\nips in the subnet cidr, and used ips in the allocation pools.\n\nThis change is for adding fields to the api result.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/965083\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/947826\n\nRelated-Bug: #2107316\nChange-Id: Ia6e264c8318636f22ae26c60589054944df66542\nSigned-off-by: Jimin Shin \u003cjimin3.shin@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/68bfa5b1faa19e75776fdb22f9323dc5108018ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/68bfa5b1faa19e75776fdb22f9323dc5108018ed"}]},"branch":"refs/heads/master"},"e4c9f024bbf13eede44c54e70a746cd3e274da5d":{"kind":"REWORK","_number":6,"created":"2025-11-11 05:50:53.000000000","uploader":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"ref":"refs/changes/87/965087/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/965087/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/6"}}},"commit":{"parents":[{"commit":"7b3be70b350c3cbe4298bc05ae648fe5fadb19b4","subject":"Fix duplicate registration for resource_extend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7b3be70b350c3cbe4298bc05ae648fe5fadb19b4"}]}],"author":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-10-24 11:17:25.000000000","tz":540},"committer":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-11-11 05:44:39.000000000","tz":540},"subject":"Add information to network ip availabilities","message":"Add information to network ip availabilities\n\nThe change[0] provides full information on the number of ips in the\nsubnet from the response of GET ``/v2.0/network-ip-availabilites``\n- the number of ips in the subnet cidr, in the allocation pools, used\nips in the subnet cidr, and used ips in the allocation pools.\n\nThis change is for adding fields to the api result.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/965083\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/947826\n\nRelated-Bug: #2107316\nChange-Id: Ia6e264c8318636f22ae26c60589054944df66542\nSigned-off-by: Jimin Shin \u003cjimin3.shin@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e4c9f024bbf13eede44c54e70a746cd3e274da5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e4c9f024bbf13eede44c54e70a746cd3e274da5d"}]},"branch":"refs/heads/master"},"6726aa5b10dd9b2e711dd16083a3f39f028a15a2":{"kind":"REWORK","_number":7,"created":"2025-12-03 05:50:28.000000000","uploader":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"ref":"refs/changes/87/965087/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/965087/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/7"}}},"commit":{"parents":[{"commit":"7b3be70b350c3cbe4298bc05ae648fe5fadb19b4","subject":"Fix duplicate registration for resource_extend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7b3be70b350c3cbe4298bc05ae648fe5fadb19b4"}]}],"author":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-10-24 11:17:25.000000000","tz":540},"committer":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-12-03 05:49:23.000000000","tz":540},"subject":"Add information to network ip availabilities","message":"Add information to network ip availabilities\n\nThe change[0] provides full information on the number of ips in the\nsubnet from the response of GET ``/v2.0/network-ip-availabilites``\n- the number of ips in the subnet cidr, in the allocation pools, used\nips in the subnet cidr, and used ips in the allocation pools.\n\nThis change is for adding fields to the api result.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/965083\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/947826\n\nRelated-Bug: #2107316\nChange-Id: Ia6e264c8318636f22ae26c60589054944df66542\nSigned-off-by: Jimin Shin \u003cjimin3.shin@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6726aa5b10dd9b2e711dd16083a3f39f028a15a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6726aa5b10dd9b2e711dd16083a3f39f028a15a2"}]},"branch":"refs/heads/master"},"17b5e536d2316d2a615d466ee7d0d2b4804bd7f1":{"kind":"REWORK","_number":8,"created":"2025-12-03 09:59:59.000000000","uploader":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"ref":"refs/changes/87/965087/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/965087/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/8"}}},"commit":{"parents":[{"commit":"7b3be70b350c3cbe4298bc05ae648fe5fadb19b4","subject":"Fix duplicate registration for resource_extend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7b3be70b350c3cbe4298bc05ae648fe5fadb19b4"}]}],"author":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-10-24 11:17:25.000000000","tz":540},"committer":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-12-03 09:59:48.000000000","tz":540},"subject":"Add information to network ip availabilities","message":"Add information to network ip availabilities\n\nThe change[0] provides full information on the number of ips in the\nsubnet from the response of GET ``/v2.0/network-ip-availabilites``\n- the number of ips in the subnet cidr, in the allocation pools, used\nips in the subnet cidr, and used ips in the allocation pools.\n\nThis change is for adding fields to the api result.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/965083\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/947826\n\nRelated-Bug: #2107316\nChange-Id: Ia6e264c8318636f22ae26c60589054944df66542\nSigned-off-by: Jimin Shin \u003cjimin3.shin@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/17b5e536d2316d2a615d466ee7d0d2b4804bd7f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/17b5e536d2316d2a615d466ee7d0d2b4804bd7f1"}]},"branch":"refs/heads/master"},"67b04fc5eaa6140a03ea36163c8bd3418def8d9e":{"kind":"REWORK","_number":9,"created":"2025-12-03 11:13:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/87/965087/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/965087/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/9"}}},"commit":{"parents":[{"commit":"7b3be70b350c3cbe4298bc05ae648fe5fadb19b4","subject":"Fix duplicate registration for resource_extend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7b3be70b350c3cbe4298bc05ae648fe5fadb19b4"}]}],"author":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-10-24 11:17:25.000000000","tz":540},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-12-03 11:13:04.000000000","tz":0},"subject":"Add information to network ip availabilities","message":"Add information to network ip availabilities\n\nThe change[0] provides full information on the number of ips in the\nsubnet from the response of GET ``/v2.0/network-ip-availabilites``\n- the number of ips in the subnet cidr, in the allocation pools, used\nips in the subnet cidr, and used ips in the allocation pools.\n\nThis change is for adding fields to the api result.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/965083\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/947826\n\nRelated-Bug: #2107316\nChange-Id: Ia6e264c8318636f22ae26c60589054944df66542\nSigned-off-by: Jimin Shin \u003cjimin3.shin@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/67b04fc5eaa6140a03ea36163c8bd3418def8d9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/67b04fc5eaa6140a03ea36163c8bd3418def8d9e"}]},"branch":"refs/heads/master"},"b1afccb688ad32de5b86038ae36ae03ffc60dce0":{"kind":"REWORK","_number":10,"created":"2025-12-04 04:24:26.000000000","uploader":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"ref":"refs/changes/87/965087/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/965087/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/10"}}},"commit":{"parents":[{"commit":"7b3be70b350c3cbe4298bc05ae648fe5fadb19b4","subject":"Fix duplicate registration for resource_extend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7b3be70b350c3cbe4298bc05ae648fe5fadb19b4"}]}],"author":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-10-24 11:17:25.000000000","tz":540},"committer":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-12-04 04:24:08.000000000","tz":540},"subject":"Add information to network ip availabilities","message":"Add information to network ip availabilities\n\nThe change[0] provides full information on the number of ips in the\nsubnet from the response of GET ``/v2.0/network-ip-availabilites``\n- the number of ips in the subnet cidr, in the allocation pools, used\nips in the subnet cidr, and used ips in the allocation pools.\n\nThis change is for adding fields to the api result.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/965083\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/947826\n\nRelated-Bug: #2107316\nChange-Id: Ia6e264c8318636f22ae26c60589054944df66542\nSigned-off-by: Jimin Shin \u003cjimin3.shin@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b1afccb688ad32de5b86038ae36ae03ffc60dce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b1afccb688ad32de5b86038ae36ae03ffc60dce0"}]},"branch":"refs/heads/master"},"b7dd86ea0c6857f48010af63941766281b5a4570":{"kind":"REWORK","_number":11,"created":"2025-12-05 06:26:30.000000000","uploader":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"ref":"refs/changes/87/965087/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/965087/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/965087/11"}}},"commit":{"parents":[{"commit":"7b3be70b350c3cbe4298bc05ae648fe5fadb19b4","subject":"Fix duplicate registration for resource_extend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7b3be70b350c3cbe4298bc05ae648fe5fadb19b4"}]}],"author":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-10-24 11:17:25.000000000","tz":540},"committer":{"name":"jimin3-shin","email":"jimin3.shin@samsung.com","date":"2025-12-05 05:25:27.000000000","tz":540},"subject":"Add information to network ip availabilities","message":"Add information to network ip availabilities\n\nThe change[0] provides full information on the number of ips in the\nsubnet from the response of GET ``/v2.0/network-ip-availabilites``\n- the number of ips in the subnet cidr, in the allocation pools, used\nips in the subnet cidr, and used ips in the allocation pools.\n\nThis change is for adding fields to the api result.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/965083\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/947826\n\nRelated-Bug: #2107316\nChange-Id: Ia6e264c8318636f22ae26c60589054944df66542\nSigned-off-by: Jimin Shin \u003cjimin3.shin@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b7dd86ea0c6857f48010af63941766281b5a4570"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b7dd86ea0c6857f48010af63941766281b5a4570"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
