)]}'
{"id":"openstack%2Fneutron-lib~965929","triplet_id":"openstack%2Fneutron-lib~master~Iec8f1a3b3094c7d75882062f006d1efe77f71ca2","project":"openstack/neutron-lib","branch":"master","topic":"bug/1875516","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-11-11 12:10:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iec8f1a3b3094c7d75882062f006d1efe77f71ca2","subject":"Remove ACCESS_READONLY RBAC constant.","status":"MERGED","created":"2025-11-03 13:20:38.000000000","updated":"2025-11-11 12:11:09.000000000","submitted":"2025-11-11 12:10:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965929-bug/1875516","meta_rev_id":"96ac9212a8d445d01fd89e03282e209f918c0b84","_number":965929,"virtual_id_number":965929,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-11 12:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-11-04 13:28:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-11-04 12:52:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-11-04 13:28:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-03 15:15:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-03 15:21:21.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-11-04 12:52:50.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-11-04 13:28:00.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-11-10 10:54:51.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"}],"messages":[{"id":"fd085895c97e29f1bc7811fd3106c6cf585c1f98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-11-03 13:20:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30243aaea26cda44d09d803bd574e15c2905a967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 15:15:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35720efaa1194b0b8d21b6fdaa096e30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce32783552964248b4f3dec75b2233f1 : SUCCESS in 8m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/873b8521af5b44b9aee0c4b8774e8355 : FAILURE in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cc25ba0daa44e36b79b1b3c92f9ac2d : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a995717fb52d4549a749c0a80a59496f : FAILURE in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b780049e0b5c4dd9a673dac4843c30bd : FAILURE in 6m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2569f945db5497e8af766dd3f15a506 : SUCCESS in 1h 17m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8f00242785f456ab180580438c18618 : SUCCESS in 3m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c023b131417b4861ac25e3971ea26e94 : SUCCESS in 1h 26m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d2642e7fdd6548acb6365be2b33590f9 : SUCCESS in 1h 44m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/78e5decb8d454c009b090fb5d21479eb : SUCCESS in 32m 04s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/7f4fa0d97ebe4c74b429cef93d893464 : SUCCESS in 30m 05s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d933200309dc43b1820636128c044da0 : FAILURE in 5m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b809c075bd3b40b9a4a9000831055d97 : SUCCESS in 1h 47m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d1427cc81febeabd54b5a6e3505f14ff287feb01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-11-04 09:28:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"94cea3de0b8fe489a20bae363a45064e683f2757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 11:35:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d1f6686d7b548c48a3d25112ce765b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33113ebfeb3d4e658565b019ad855d93 : SUCCESS in 9m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9e39c14ae7a4bac95d9ffd2615a4f17 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2bbdcb4012845b287c6138c8157740c : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/586a561328ea4da09048b937b1759db4 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d7538c98cce47d4a37ef112b5d5ec80 : SUCCESS in 5m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f10b5f010eef4503ba6cf30cb8c3212e : SUCCESS in 1h 43m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd38c17e26d84f8dbcf6f3eedc04ab36 : SUCCESS in 4m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/28dd02cad6c24caabb6f3c7874bed34a : SUCCESS in 1h 34m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3fa65f22958d4869958e3ae46025c645 : SUCCESS in 2h 04m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8dbf803f10364ecabe0d9b309e01b14f : SUCCESS in 36m 05s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/09d3f859e4ac45fbbda434cb686bbe4d : SUCCESS in 23m 47s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/4d00432e6165422a96b68a3488c047cc : SUCCESS in 4m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e871244af0b14a1b88c84364fbf9801e : SUCCESS in 1h 36m 07s","accounts_in_message":[],"_revision_number":2},{"id":"74b994e5ce396fd02389c89d089f6891af440a87","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-04 12:52:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"eb27204d60dc2ff2fbfd9b6f41d3842a918f49a6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-04 13:28:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"da5233fed525606031320c975ba6feefd9bb76b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 13:28:03.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61175b5bbf0345a285e0ead28f9b0f28\n\n\nWarning:\n  Change 965928,2 in project openstack/neutron does not share a change queue with 965929 in project openstack/neutron-lib","accounts_in_message":[],"_revision_number":2},{"id":"5887d68cafe4bc44213023936fdeade57ebbc576","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-10 10:54:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"85bb74423b4c7761cf72efb3033c09131ca351af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 12:44:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34598a324a1e4f73b9dd09cdadb779cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31501ba0ace34203a5453c9ee6aecd5d : SUCCESS in 6m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62b1e32946364f78a922b4b65a4268cd : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5028f2a4622045beb563674b7ef602af : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c11edfa11f84781bc82650f4e6a9c34 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/843009026a7041f78f20db2db6e1a40e : SUCCESS in 6m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a24360e35f249509be02716ad5cc867 : SUCCESS in 1h 39m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/865ac4b29ea445238f2fe72d604f0ea0 : SUCCESS in 3m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3986e08135714a898e87aeee3feca64e : SUCCESS in 1h 48m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5bc0b65d83954347a2d04c6c2d163162 : SUCCESS in 1h 46m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/90f66c2b51d8462e9a402bbfcc95150e : SUCCESS in 32m 04s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/67f46982373c4eae87d348921e101fdc : SUCCESS in 32m 58s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/1aba516fd32e4e28b95834f25aad7e8c : SUCCESS in 5m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6cfd2cbc480d4f7693a286c17b2fec36 : FAILURE in 49m 43s","accounts_in_message":[],"_revision_number":2},{"id":"dbccb0c1d8645abab78dcb5a487366f547e659ed","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-11 07:58:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4e18b9aa53cfbb247f2016a12ab00762d3adf47c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 10:09:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3be18fb5ff1b47a3a398fadf7683aa7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72da305f9186478090b1b24753b545f0 : SUCCESS in 8m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9a187829453412db1c63d6fed5f35be : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d81099f489a2407fb53293c733441549 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6db3bf24390d4309867b9682ba811b9f : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56768fdb72e2457f94ba623b50424295 : SUCCESS in 5m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9c4acd306304b84968646841bc2f645 : SUCCESS in 2h 04m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fda5b4306894719b7ebb3e1a24be7b4 : SUCCESS in 3m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6498f58bd381458eb4376c4ccf567da2 : SUCCESS in 1h 20m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cc21a995ea384575a45f6677dcc9ca98 : SUCCESS in 2h 07m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c1ee627dc5fd4bc1a649b817b381e709 : SUCCESS in 32m 08s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/063f0e64ead9400da98a911705c645dd : SUCCESS in 28m 52s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3a6d051474c3452ca081fc53931df4b5 : SUCCESS in 6m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b883d1b95eaf40efab1880eeadd307bf : SUCCESS in 1h 43m 26s","accounts_in_message":[],"_revision_number":2},{"id":"c8959ab7ca1eda7931d759c8a5c7b08d507b5d7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 10:10:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"47e715210e8cd40637b269cc25b8fcb0c0caa428","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 12:10:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8aed5a37fed4ecda5ed5c29c2691659\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c8b948cf7df4acea936ee20da4f8b82 : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6154e8e898974ddb80795d0bae5b0c6b : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/784daf2aedb9489da43c72310b7714a5 : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/82bd9a5c67d44dd8840c6618a5358f26 : SUCCESS in 4m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a99dc88be57c4b8db23d064290c0bdab : SUCCESS in 1h 54m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91db5b16d6654ca683915c17720331f5 : SUCCESS in 3m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d797efbb9c204dfa9749666f1a51468c : SUCCESS in 1h 21m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/18957c39477a484c851d0ad2848854bd : SUCCESS in 1h 59m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/971189248b1f4dc78c1dbd5a334afd42 : SUCCESS in 1h 44m 19s","accounts_in_message":[],"_revision_number":2},{"id":"9d96cfa254a1c59ee9df57b4144d1a3924fcd7be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 12:10:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"96ac9212a8d445d01fd89e03282e209f918c0b84","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 12:11:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10b0e92e0ce54dd681812481ec2376e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c5cdb203ea94bf3a364d4f8444313ac : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b69557e6d1b6466590f97c87381fa2d3 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7f4285b96dc5d251accb7a0e24fdcb2fbc8c2e41","revisions":{"4b6ecf658ddd44c8e6f8cb4b72431a776d243a70":{"kind":"REWORK","_number":1,"created":"2025-11-03 13:20:38.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/29/965929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/29/965929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/965929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/965929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/965929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/29/965929/1"}}},"commit":{"parents":[{"commit":"ba498831235681f2d996d1d8b0bf9b7b53800f0e","subject":"reno: Update master for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ba498831235681f2d996d1d8b0bf9b7b53800f0e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-11-03 13:11:12.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-11-03 13:20:36.000000000","tz":60},"subject":"Remove ACCESS_READONLY RBAC constant.","message":"Remove ACCESS_READONLY RBAC constant.\n\nThis partially reverts commit [1] and removes constant ACCESS_READONLY\nfrom the neutron_lib.constants module.\nThat constant was introduced as part of the RFE\nhttps://bugs.launchpad.net/neutron/+bug/1875516 which was never fully\nimplemented and currently, with new S-RBAC API policies is not really\nneeded anymore.\nIt was decided during the Gazpacho PTG that this RFE will be closed and\nalready merged bits will be cleaned. That patch cleans it from the\nneutron-lib repo.\n\n[1] https://review.opendev.org/c/openstack/neutron-lib/+/730736\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/965928\n\nRelated-bug: #1875516\n\nChange-Id: Iec8f1a3b3094c7d75882062f006d1efe77f71ca2\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4b6ecf658ddd44c8e6f8cb4b72431a776d243a70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4b6ecf658ddd44c8e6f8cb4b72431a776d243a70"}]},"branch":"refs/heads/master"},"7f4285b96dc5d251accb7a0e24fdcb2fbc8c2e41":{"kind":"REWORK","_number":2,"created":"2025-11-04 09:28:59.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/29/965929/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/29/965929/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/965929/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/965929/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/965929/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/29/965929/2"}}},"commit":{"parents":[{"commit":"ba498831235681f2d996d1d8b0bf9b7b53800f0e","subject":"reno: Update master for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ba498831235681f2d996d1d8b0bf9b7b53800f0e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-11-03 13:11:12.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-11-04 09:28:51.000000000","tz":60},"subject":"Remove ACCESS_READONLY RBAC constant.","message":"Remove ACCESS_READONLY RBAC constant.\n\nThis partially reverts commit [1] and removes constant ACCESS_READONLY\nfrom the neutron_lib.constants module.\nThat constant was introduced as part of the RFE\nhttps://bugs.launchpad.net/neutron/+bug/1875516 which was never fully\nimplemented and currently, with new S-RBAC API policies is not really\nneeded anymore.\nIt was decided during the Gazpacho PTG that this RFE will be closed and\nalready merged bits will be cleaned. That patch cleans it from the\nneutron-lib repo.\n\n[1] https://review.opendev.org/c/openstack/neutron-lib/+/730736\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/965928\n\nRelated-bug: #1875516\n\nChange-Id: Iec8f1a3b3094c7d75882062f006d1efe77f71ca2\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7f4285b96dc5d251accb7a0e24fdcb2fbc8c2e41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7f4285b96dc5d251accb7a0e24fdcb2fbc8c2e41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
