)]}'
{"id":"openstack%2Fneutron-lib~966419","triplet_id":"openstack%2Fneutron-lib~master~If4c0fef689be8d1f78a0822c1a50d700709860db","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"13039":{"account":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"last_update":"2025-11-10 12:05:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4c0fef689be8d1f78a0822c1a50d700709860db","subject":"Fix documentation for maximum tag length","status":"MERGED","created":"2025-11-07 11:21:29.000000000","updated":"2025-11-10 12:06:29.000000000","submitted":"2025-11-10 12:05:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966419","meta_rev_id":"84f8714832d36e2492bd78aebc222c8d0b3908b9","_number":966419,"virtual_id_number":966419,"owner":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-10 12:05:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-11-10 10:18:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-11-07 13:36:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2025-11-10 10:18:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-07 13:36:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-11-07 13:58:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-10 10:18:02.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"a95872a010cbbb8f675e2d5233ee562056b24bd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2025-11-07 11:21:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb4c715bea709e5b5f48be25e43c3f37f7343fb2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-07 13:36:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1c8c1f087773281c4ef4347b5aff915f0b09b6f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 13:58:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/485f7aac7ebc48e59191f82309de1ed5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0efec036b0a54591abdbfea69cbf5ab4 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21afeba3afc440f0ad16113e85e5ba10 : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/784655af8a3b405982aa14822dc9967e : SUCCESS in 3m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/95680cf05d8e49828359349de63ee66a : SUCCESS in 1h 32m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/71ef71d441fa4c199d38ca8db8ab7af8 : SUCCESS in 2h 35m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d6dd9c2a54054dbf976a2ed42f8a1a96 : SUCCESS in 31m 32s","accounts_in_message":[],"_revision_number":1},{"id":"2437f4bb49575c3dd5a83db7596a6b256d2f8b86","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-10 10:18:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ed6ce8cdf013f1bf2cc8cee3ac05a159ca83f80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 10:18:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0546c8195c3ce12be9385bfd95beaf2445e5ea95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 12:05:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80ab6a33c9ac401c98093f7ce962883a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/545488fe512c45d0ac052a755626b237 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e3250a185a46feb3e3dce40b98769a : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b957e2a762674ace8b0ed8f2bb2e9334 : SUCCESS in 4m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8cc5fda4ed4543a097ca65877282c950 : SUCCESS in 1h 31m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4475b1b574d341ca8a79fabc8f740e54 : SUCCESS in 1h 46m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a8895ffa809e98ce055b89d2e46d35cfed6a7a79","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 12:05:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"84f8714832d36e2492bd78aebc222c8d0b3908b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 12:06:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fc2a90f30df4d0984c7e277690d2ad5\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/280b477073304f228afcaf98b20a0cc7 : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3545daf34da3476b882d04bb72c7660f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2feaec5a559b463a53739062668d7033ae8376b7","revisions":{"2feaec5a559b463a53739062668d7033ae8376b7":{"kind":"REWORK","_number":1,"created":"2025-11-07 11:21:29.000000000","uploader":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"ref":"refs/changes/19/966419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/19/966419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/19/966419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/19/966419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/19/966419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/19/966419/1"}}},"commit":{"parents":[{"commit":"ba498831235681f2d996d1d8b0bf9b7b53800f0e","subject":"reno: Update master for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ba498831235681f2d996d1d8b0bf9b7b53800f0e"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2025-11-07 11:14:50.000000000","tz":60},"committer":{"name":"Martin André","email":"m.andre@redhat.com","date":"2025-11-07 11:21:16.000000000","tz":60},"subject":"Fix documentation for maximum tag length","message":"Fix documentation for maximum tag length\n\nIt was changed from 60 to 255 via\nIe69526acb94b62fd5d8db1dbddc1f24072df7a5e but the doc was never updated\nto reflect the change.\n\nChange-Id: If4c0fef689be8d1f78a0822c1a50d700709860db\nSigned-off-by: Martin André \u003cm.andre@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2feaec5a559b463a53739062668d7033ae8376b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2feaec5a559b463a53739062668d7033ae8376b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
