)]}'
{"id":"openstack%2Fneutron-lib~969952","triplet_id":"openstack%2Fneutron-lib~master~I1d2377fb452896544573b1aabb70df0623d9e097","project":"openstack/neutron-lib","branch":"master","topic":"bug/2133212","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-12-16 09:06:58.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-12-16 11:12:26.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-12-10 14:11:03.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I1d2377fb452896544573b1aabb70df0623d9e097","subject":"Deprecate \"is_advsvc\" attribute in the Context object","status":"MERGED","created":"2025-12-05 15:25:21.000000000","updated":"2025-12-16 11:13:23.000000000","submitted":"2025-12-16 11:12:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969952-bug/2133212","meta_rev_id":"49491c085b7479cb79e7951d14e7d068d75737b6","_number":969952,"virtual_id_number":969952,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-16 11:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-12-16 09:06:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-12-10 14:11:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-12-16 09:06:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-05 17:11:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-12-05 17:30:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-05 21:58:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-12-09 07:55:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-12-09 17:41:46.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"af06f60fd84a7cf8d3f6cb3478199430e5719933","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-05 15:25:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9177a9f0b2e612869c1a9592638bdab3eb17bfc0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-05 17:11:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2851301d3f56f4c8125295e5f92016ad4c13d5a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 17:30:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b6396990f844248932d7218025ebc2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c281bd5cc194475db71b222b21bf7319 : SUCCESS in 9m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/041062f7c3c44373aa413131169e3ffb : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f93975260b3041ba8527e130b2c064a0 : SUCCESS in 6m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e79919be30e4529a8318f64a3d5fd14 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae56c936a4ce443fb4a20cb0fc95cba5 : SUCCESS in 3m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8790b444e384490595b5028a05526287 : FAILURE in 42m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1aca52e636064dcfa9f0757df8ff2e4c : SUCCESS in 1h 26m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d8daeb02753c41d0b4c1429c9d1b31a4 : SUCCESS in 2h 01m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/72174eaaf5264e6c902627992219efd7 : SUCCESS in 33m 34s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/21a8fea9652c4c478d3b82cf0a3a904e : SUCCESS in 29m 38s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6879b91399c94942b350a5f8f8309342 : SUCCESS in 6m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a6f2661b4f9b44aeb5004e2f916ee0eb : SUCCESS in 1h 58m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4e58c2f1930b6e6e4dd07beacc3bb01caeafb6df","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-05 21:58:45.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0df5497e67499597e8cf3532d193b833e0d553ef","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-09 07:55:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ae0b4f633eaaed7108fba67d63b6c94c80864c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-09 09:25:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"afd47ec68ce874c704b8cf9ae984af5dd2b32667","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-09 09:25:58.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dc8f6efdaead410fcc98c546df224fc025175734","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-09 10:45:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b6fb0431b86b3d149fd7c0277d9716ebf997aea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 11:21:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3aa633dc46e74fc5a2ec9df1a308bb80\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15245f8b2c844fb3b683c51281433262 : SUCCESS in 8m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70adb7d134654bcc918b446ade328311 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dda883a9f0447f4917e1b4fe5037b7d : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08e7be961b7845feb8f70548ee9ae88c : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ff79e947e804566b446bfc571e381bf : SUCCESS in 4m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6032a042a0424517a5f36d05df888481 : SUCCESS in 1h 43m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aef17a4283fa4e93ab54f02c5cf52f5c : SUCCESS in 4m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/60e2586c9e084761a310c673a3be4ab6 : SUCCESS in 1h 41m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e78a7246c07548a89bb0f2b308f83f8c : SUCCESS in 1h 48m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6cafa2868a984d5eada30807f7829e1a : SUCCESS in 32m 39s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/53b00a3bb28d4219897aa4baad63ba98 : SUCCESS in 26m 00s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/56891580a0e241b491bc22a4af3d253e : SUCCESS in 5m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c7dea279c4804c17bf55a04833a18ced : SUCCESS in 1h 50m 03s","accounts_in_message":[],"_revision_number":2},{"id":"eca00d39b45952da4915fef6e88c501e064d6560","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-09 14:27:20.000000000","message":"Topic I1d2377fb452896544573b1aabb70df0623d9e097 removed","accounts_in_message":[],"_revision_number":2},{"id":"6b84f11351464fb175f2d0a87ca68f31fd8146d8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-09 14:27:25.000000000","message":"Topic set to bug/2133212","accounts_in_message":[],"_revision_number":2},{"id":"ccb868173ff4aa9eb2442119795f4d6440e149d8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-09 17:41:46.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7086555955afacd269bc53519572d660c12d4c83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-10 13:02:56.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2cde9d0a0b17e7829d1d2a44f14b7b4b04821892","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-10 13:03:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6c10093dbc3eb74cdadef6cf5c771c9624b940ed","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-10 14:11:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"50d6b1603b0b34aecc24762af7e35b90f595179b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 15:44:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f9430ace05b4a28bd2400fa6507e167\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/628305bb9b7d498893a4ca980ef510db : SUCCESS in 6m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/793d65cc9fa94f6cba96ec4d02982c66 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9711ec1b04894330b78fafce06240b43 : SUCCESS in 8m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d4e557b5c5e42a9b5966870b7f7109c : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d533c906f59a4c28b0414ae61908a165 : SUCCESS in 3m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/532f6bf4ce21480e9574c2dac25844a7 : SUCCESS in 1h 44m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/562aa206c0f34ced88e2fca845c06844 : SUCCESS in 3m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/41e71d3d312043e19f851402e5a85303 : SUCCESS in 1h 34m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/efad530de77740afaa8fae9366d93cbb : SUCCESS in 2h 40m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cf0795d34a744fd28db64a68ea20f00f : SUCCESS in 35m 45s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/ceccf5a01aab42d296da0d9bb74d4977 : SUCCESS in 31m 37s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3a1cb03d7f0943449f23b872d3173998 : SUCCESS in 5m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0e25861c224f4614b9e1cacc7019e0b0 : SUCCESS in 1h 41m 23s","accounts_in_message":[],"_revision_number":3},{"id":"a8f65e8732d085ad70a7ddf8722b1647b4daad6d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-16 09:06:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9ec42b2190b4cbaeacf81402782c485f93407067","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 09:07:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cd9f9a9b91f4654b525f178b04c583d72aa6804c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 11:12:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a300004c10747ef80686863217aff99\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78f12b192ef1413cb5aee52db81fe264 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cb38f3ded87410696352f534656c0ee : SUCCESS in 8m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e45a72081d564378a0123cd12bfca698 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1c4fb0d23a1d4538ba0a4d1b61244cd3 : SUCCESS in 3m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99bd3f8d3faa4b75834a91f68bc1f12f : SUCCESS in 56m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60d4cc51c0b24a68ad12e366192e1914 : SUCCESS in 4m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fe310b85cc45436f8e809dd6e5b7549c : SUCCESS in 1h 19m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/43af5130ff5b43798069dbb653df67a9 : SUCCESS in 2h 04m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/026bbf4cada34973959a51e6efd782fb : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":3},{"id":"c489809f8850e47cf3370f073aef625177dc0518","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 11:12:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"49491c085b7479cb79e7951d14e7d068d75737b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 11:13:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e66c3a4b63c945f4a85930e2fe26457d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c583ed589f64472cb4f00b9cdc939636 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6435f9521d6349aa9310b042cc1210e5 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b5272aabdabb1776e67dac6f517c2c52331ac293","revisions":{"8aac3c44e82bd5424132c5900fcd37a7ed3ec3b7":{"kind":"REWORK","_number":1,"created":"2025-12-05 15:25:21.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/52/969952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/969952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/969952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/969952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/969952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/969952/1"}}},"commit":{"parents":[{"commit":"20ab081c38d28359a8adf0c40c37ce800897a4e6","subject":"Fix assertion in unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/20ab081c38d28359a8adf0c40c37ce800897a4e6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-05 15:04:47.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-05 15:10:13.000000000","tz":60},"subject":"Deprecate \"is_advsvc\" attribute in the Context object","message":"Deprecate \"is_advsvc\" attribute in the Context object\n\nSince some time we have already \"context.Context.is_service_role\"\nattribute in the context.Context object and this one should be used\ninstead of old \"is_advsvc\" with S-RBAC.\nThere was however missing __init__ method attribute to force configure\nContext as service_role context - this was possible until now only to do\nwith is_advsvc parameter.\n\nThis patch adds is_service_role parameter to the\ncontext.Context.__init__ method and deprecates old attribute\n\"is_advsvc\".\n\nChange-Id: I1d2377fb452896544573b1aabb70df0623d9e097\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8aac3c44e82bd5424132c5900fcd37a7ed3ec3b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8aac3c44e82bd5424132c5900fcd37a7ed3ec3b7"}]},"branch":"refs/heads/master"},"a3d6b79f35025682a1daa6339acab64cea0cd79a":{"kind":"REWORK","_number":2,"created":"2025-12-09 09:25:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/52/969952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/969952/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/969952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/969952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/969952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/969952/2"}}},"commit":{"parents":[{"commit":"20ab081c38d28359a8adf0c40c37ce800897a4e6","subject":"Fix assertion in unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/20ab081c38d28359a8adf0c40c37ce800897a4e6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-05 15:04:47.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-09 09:25:34.000000000","tz":60},"subject":"Deprecate \"is_advsvc\" attribute in the Context object","message":"Deprecate \"is_advsvc\" attribute in the Context object\n\nSince some time we have already \"context.Context.is_service_role\"\nattribute in the context.Context object and this one should be used\ninstead of old \"is_advsvc\" with S-RBAC.\nThere was however missing __init__ method attribute to force configure\nContext as service_role context - this was possible until now only to do\nwith is_advsvc parameter.\n\nThis patch adds is_service_role parameter to the\ncontext.Context.__init__ method and deprecates old attribute\n\"is_advsvc\".\n\nChange-Id: I1d2377fb452896544573b1aabb70df0623d9e097\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a3d6b79f35025682a1daa6339acab64cea0cd79a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a3d6b79f35025682a1daa6339acab64cea0cd79a"}]},"branch":"refs/heads/master"},"b5272aabdabb1776e67dac6f517c2c52331ac293":{"kind":"REWORK","_number":3,"created":"2025-12-10 13:02:56.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/52/969952/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/969952/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/969952/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/969952/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/969952/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/969952/3"}}},"commit":{"parents":[{"commit":"20ab081c38d28359a8adf0c40c37ce800897a4e6","subject":"Fix assertion in unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/20ab081c38d28359a8adf0c40c37ce800897a4e6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-05 15:04:47.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-10 13:02:56.000000000","tz":0},"subject":"Deprecate \"is_advsvc\" attribute in the Context object","message":"Deprecate \"is_advsvc\" attribute in the Context object\n\nSince some time we have already \"context.Context.is_service_role\"\nattribute in the context.Context object and this one should be used\ninstead of old \"is_advsvc\" with S-RBAC.\nThere was however missing __init__ method attribute to force configure\nContext as service_role context - this was possible until now only to do\nwith is_advsvc parameter.\n\nThis patch adds is_service_role parameter to the\ncontext.Context.__init__ method and deprecates old attribute\n\"is_advsvc\".\n\nChange-Id: I1d2377fb452896544573b1aabb70df0623d9e097\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b5272aabdabb1776e67dac6f517c2c52331ac293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b5272aabdabb1776e67dac6f517c2c52331ac293"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
