)]}'
{"id":"openstack%2Fneutron-lib~969962","triplet_id":"openstack%2Fneutron-lib~master~I6328c5977488066bcb8247111a7a629b88fd487a","project":"openstack/neutron-lib","branch":"master","topic":"2134096","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-12-17 17:12:43.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-12-17 14:41:14.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-12-17 17:12:43.000000000","reason":"Change was submitted"},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2025-12-17 17:12:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6328c5977488066bcb8247111a7a629b88fd487a","subject":"Add \"both\" direction to tap mirror definition","status":"MERGED","created":"2025-12-05 16:10:13.000000000","updated":"2025-12-17 17:13:44.000000000","submitted":"2025-12-17 17:12:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969962-2134096","meta_rev_id":"32682cbbf8d0ae994a9290b14a944d7317108beb","_number":969962,"virtual_id_number":969962,"owner":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-17 17:12:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-12-17 14:19:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-12-17 14:41:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-12-17 14:41:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-05 18:10:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-08 08:54:21.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-12-09 08:17:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-12-09 14:22:26.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-12-10 16:35:29.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"}],"messages":[{"id":"f3dbeea4055be55312df04230ebf57042c9cc958","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2025-12-05 16:10:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78d06d4a552c97dd08f1d2a81d1780405d0d45a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:10:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e1335900f664cddb2d8119c3f7fabac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88d001ee38f24e3299bfec011f267615 : SUCCESS in 4m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2de18dff79134fd0a1249414e7b84b27 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13fc51736514401aa78e50eb871ffc8e : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/994a488c29504b9c9dd8783b5ed7b63f : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ff4d63bb6634230a7e2de5932bfb5e3 : SUCCESS in 6m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6dc930b804dd4452851ba73890f81463 : SUCCESS in 1h 54m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/643fd96a44314812a45df6aa6ffdef10 : SUCCESS in 1h 19m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9bcadd29c5ea42f3aa05c7b61d6c2e07 : SUCCESS in 1h 52m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/10e7f7f4d86046fa9de047cc86812522 : SUCCESS in 31m 04s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/8502615dda734161997b7bca78a10b8a : SUCCESS in 20m 51s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d48c74094660465b849e5d04e0bc5fdb : SUCCESS in 4m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/359e9bf7c85b4bbf87b415902b9ced62 : SUCCESS in 1h 44m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b9e6c6f71a14a6fecbdc62ab21f1b5a239bf0d74","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-08 08:54:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"155395330522398bce680ba1dc7f67a54c3cbd64","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-09 08:17:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"348569f2917724b2f542e3039f2fae32a46476ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2025-12-16 14:08:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"bdd54c72d9219c2203ab2253a291960d006ad660","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-16 14:52:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"dd02db853121029e6746b641d4bb1a52274546ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2025-12-16 16:06:20.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"edb460c17d39419bba91ba3ef57c40d28053e61e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2025-12-16 16:11:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"150e298ce2bbc94657c2d25ff9fd8452b9ee15b0","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2025-12-16 16:14:40.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b16acef9bcbb7d467a0bba2b4ff40057dcad17fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 18:25:27.000000000","message":"Patch Set 4: Verified-1\n\n(23 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/905c0f299b444dc3a614315b74618a1a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/083935ba3097485591428a27c3ab90a4 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0fb23bd62af4a21ac1f348fd2202604 : SUCCESS in 9m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7ccea15bd7547bca0c5b534d880ad5b : FAILURE in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6e4d66c40b34707b99a1d9ac57308bb : FAILURE in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/048c002052a7428bb4ccc937898bbce3 : FAILURE in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68b35770da5742c2983f8864e29ec1fe : FAILURE in 4m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1b925aa4e5a4cbbbd84af39e8cd06f3 : SUCCESS in 1h 31m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a9b0bec24b44ffd893c7f3860417e0b : FAILURE in 1m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/574571b0af9347f3a7c7afa5f5d9e8e3 : SUCCESS in 1h 24m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3fc1ee6b70864554a707e1d00e9ac398 : SUCCESS in 2h 13m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b4f43a5cdf1e423993098f9e35b5b9ca : SUCCESS in 33m 24s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/9d5b06073321498cb3336091168205a3 : SUCCESS in 18m 49s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0755e15050824cb18f89d6a616559dca : FAILURE in 6m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3e00289af68a4d44908771d3625b145b : SUCCESS in 1h 04m 51s","accounts_in_message":[],"_revision_number":4},{"id":"df9608e736a7ec9d8aa484b6a01b0a6c9e316050","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-16 22:34:06.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"44442b913a6c0bdbfe915f12c05c3ed3054caa8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2025-12-17 11:39:18.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"39425c6505f6f076cd21fb0206c0e272fd6be175","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2025-12-17 12:22:03.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"15ca0a2d0ef7f37e0c45f68e00a9259d2d5c0b3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2025-12-17 12:40:35.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"79f0ca581842a64d0424161b004f733fc3f06d16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-17 12:47:42.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"8fb85ed546703f98186d076012a23e2cc7c1363b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-17 14:19:04.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"030c225812b39575abe4c429bc86dc80639ab81e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-17 14:41:14.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"e5a0b54d15d2745322952e5f3d2054696728e5c6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-17 14:41:41.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"b7829a12040b38041937e167c1d23b1507dadde5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 15:10:24.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16b93c1df4cc427283bc7d8b7a8a0372\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ebf1c9ea6cdb4bba915eeafdd1dce910 : SUCCESS in 2m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10164a74d1fb438099d9c06de827edd4 : SUCCESS in 3m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf42435dfc8a44f5adb0e5dd70ec2d85 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43c8733e550542f49c21d0cae803091e : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d356ac69ec448039f12016e50149733 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a59bf5dcb514d6c867beea0482df0ee : SUCCESS in 7m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c98d8b6fc72a4da3889b104a604cc3e2 : SUCCESS in 2h 07m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be9f65b78109489eb0474446b2e21cf1 : SUCCESS in 4m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fedef328f49647c48110ad25abebccb8 : SUCCESS in 1h 29m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a387505d52f5463f8feb2c13e6d49421 : SUCCESS in 1h 59m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6028c0410c794762b5d261173d9ac01c : SUCCESS in 31m 46s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/a133cd814dba4164902623d9645f49b8 : SUCCESS in 21m 09s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c692eb02df8944df86eea20d11d4567b : SUCCESS in 5m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/878cc582504a442986bec550b34259ad : SUCCESS in 2h 13m 29s","accounts_in_message":[],"_revision_number":8},{"id":"9f4a623d77aa4f864929527f88b2d23714eedfe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 15:10:53.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"029b5f4261e399b6910fafb9a20f7909a4c6a6c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 17:12:42.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/473d2c94e45644d48669a29ab87603c6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8510f10bd7db492883dc3eb5f26d640e : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d58bf4e64b34fae863b2c0a669b034c : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/769d825e9efb427dbd6b3e7b690437dd : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6be7a464b6d4856840fa4a220da93cd : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b2f86e9e5ad43ae963913c4b32a20e4 : SUCCESS in 5m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/877968a293404595aff6626f98935ef4 : SUCCESS in 1h 51m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c79edfbdc2354b46a881f567fe1c3a64 : SUCCESS in 5m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8c788b0d85ec460c9ce6f3c457376c7e : SUCCESS in 1h 34m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b119037a75384fb8a59c5aedfa2f1394 : SUCCESS in 2h 00m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8ed6bdc137744619b165dc13d732204f : SUCCESS in 1h 55m 37s","accounts_in_message":[],"_revision_number":8},{"id":"69e9897ddf0fddb94bdc306d71e5754c310c69d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 17:12:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"32682cbbf8d0ae994a9290b14a944d7317108beb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 17:13:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8ef9d5c1e03409a961360eebad85eff\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/b22b38457a9a4266b012513c72bd8c3a : SUCCESS in 44s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ed7c60cfc904452a9bb161f255a6f34 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f15cd60be3004a479454fa8122f6cabb : SUCCESS in 48s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"67854d125f15b743b27556f1eff6dbd9d34dbfe5","revisions":{"8677e12c08f746fbaa9ff65f0e4076967ce0c23f":{"kind":"REWORK","_number":1,"created":"2025-12-05 16:10:13.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/62/969962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/62/969962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/1"}}},"commit":{"parents":[{"commit":"950f0f3a8a722da115c03f2abb756a732a3ec4c1","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/950f0f3a8a722da115c03f2abb756a732a3ec4c1"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2025-12-05 16:07:59.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2025-12-05 16:10:11.000000000","tz":60},"subject":"Add both direction to tap mirror definition","message":"Add both direction to tap mirror definition\n\nPreviously, only IN and OUT directions calls were supported. Now\nit also supports the BOTH direction.\n\nRelated-Bug: 2134096\nChange-Id: I6328c5977488066bcb8247111a7a629b88fd487a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8677e12c08f746fbaa9ff65f0e4076967ce0c23f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8677e12c08f746fbaa9ff65f0e4076967ce0c23f"}]},"branch":"refs/heads/master"},"0bf6bd9ec62181b58163dbc41f2631eca7c6964e":{"kind":"REWORK","_number":2,"created":"2025-12-16 14:08:14.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/62/969962/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/62/969962/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/2"}}},"commit":{"parents":[{"commit":"950f0f3a8a722da115c03f2abb756a732a3ec4c1","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/950f0f3a8a722da115c03f2abb756a732a3ec4c1"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2025-12-05 16:07:59.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2025-12-16 14:07:58.000000000","tz":60},"subject":"Add \"both\" direction to tap mirror definition","message":"Add \"both\" direction to tap mirror definition\n\nPreviously, only IN and OUT directions calls were supported. Now\nit also supports the BOTH direction.\n\nRelated-Bug: 2134096\nChange-Id: I6328c5977488066bcb8247111a7a629b88fd487a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0bf6bd9ec62181b58163dbc41f2631eca7c6964e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0bf6bd9ec62181b58163dbc41f2631eca7c6964e"}]},"branch":"refs/heads/master"},"b60c1afa4d4c58783b44ca5505daa79573d835c2":{"kind":"REWORK","_number":3,"created":"2025-12-16 16:06:20.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/62/969962/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/62/969962/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/3"}}},"commit":{"parents":[{"commit":"950f0f3a8a722da115c03f2abb756a732a3ec4c1","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/950f0f3a8a722da115c03f2abb756a732a3ec4c1"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2025-12-05 16:07:59.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2025-12-16 16:06:06.000000000","tz":60},"subject":"Add \"both\" direction to tap mirror definition","message":"Add \"both\" direction to tap mirror definition\n\nPreviously, only IN and OUT directions calls were supported. Now\nit also supports the BOTH direction.\n\nRelated-Bug: 2134096\nChange-Id: I6328c5977488066bcb8247111a7a629b88fd487a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b60c1afa4d4c58783b44ca5505daa79573d835c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b60c1afa4d4c58783b44ca5505daa79573d835c2"}]},"branch":"refs/heads/master"},"8470dd8e1f746c4a6d9093c55c274878fb1970f3":{"kind":"REWORK","_number":4,"created":"2025-12-16 16:11:02.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/62/969962/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/62/969962/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/4"}}},"commit":{"parents":[{"commit":"950f0f3a8a722da115c03f2abb756a732a3ec4c1","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/950f0f3a8a722da115c03f2abb756a732a3ec4c1"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2025-12-05 16:07:59.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2025-12-16 16:10:54.000000000","tz":60},"subject":"Add \"both\" direction to tap mirror definition","message":"Add \"both\" direction to tap mirror definition\n\nPreviously, only IN and OUT directions calls were supported. Now\nit also supports the BOTH direction.\n\nRelated-Bug: 2134096\nChange-Id: I6328c5977488066bcb8247111a7a629b88fd487a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8470dd8e1f746c4a6d9093c55c274878fb1970f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8470dd8e1f746c4a6d9093c55c274878fb1970f3"}]},"branch":"refs/heads/master"},"3643ec579c711177f9fdbc0893f16c6a05f72c9e":{"kind":"REWORK","_number":5,"created":"2025-12-17 11:39:18.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/62/969962/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/62/969962/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/5"}}},"commit":{"parents":[{"commit":"950f0f3a8a722da115c03f2abb756a732a3ec4c1","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/950f0f3a8a722da115c03f2abb756a732a3ec4c1"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2025-12-05 16:07:59.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2025-12-17 11:38:23.000000000","tz":60},"subject":"Add \"both\" direction to tap mirror definition","message":"Add \"both\" direction to tap mirror definition\n\nPreviously, only IN and OUT directions calls were supported. Now\nit also supports the BOTH direction.\n\nRelated-Bug: 2134096\nChange-Id: I6328c5977488066bcb8247111a7a629b88fd487a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3643ec579c711177f9fdbc0893f16c6a05f72c9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3643ec579c711177f9fdbc0893f16c6a05f72c9e"}]},"branch":"refs/heads/master"},"b8d000049aa9874b7b1c0a6267bc65120ba805ed":{"kind":"REWORK","_number":6,"created":"2025-12-17 12:22:03.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/62/969962/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/62/969962/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/6"}}},"commit":{"parents":[{"commit":"61563e266f8f01771bb4f9a7ad60f604f2d75625","subject":"Fix tap_mirror extension name in KNOWN_EXTENSIONS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/61563e266f8f01771bb4f9a7ad60f604f2d75625"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2025-12-05 16:07:59.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2025-12-17 12:21:50.000000000","tz":60},"subject":"Add \"both\" direction to tap mirror definition","message":"Add \"both\" direction to tap mirror definition\n\nPreviously, only IN and OUT directions calls were supported. Now\nit also supports the BOTH direction.\n\nRelated-Bug: 2134096\nChange-Id: I6328c5977488066bcb8247111a7a629b88fd487a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b8d000049aa9874b7b1c0a6267bc65120ba805ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b8d000049aa9874b7b1c0a6267bc65120ba805ed"}]},"branch":"refs/heads/master"},"d09a9432fcb41834982a88ac5e839c5561bf31cc":{"kind":"REWORK","_number":7,"created":"2025-12-17 12:40:35.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/62/969962/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/62/969962/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/7"}}},"commit":{"parents":[{"commit":"28a372841402381245d65cca79c171074c4f5d21","subject":"Fix tap_mirror extension name in KNOWN_EXTENSIONS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/28a372841402381245d65cca79c171074c4f5d21"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2025-12-05 16:07:59.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2025-12-17 12:40:21.000000000","tz":60},"subject":"Add \"both\" direction to tap mirror definition","message":"Add \"both\" direction to tap mirror definition\n\nPreviously, only IN and OUT directions calls were supported. Now\nit also supports the BOTH direction.\n\nRelated-Bug: 2134096\nChange-Id: I6328c5977488066bcb8247111a7a629b88fd487a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d09a9432fcb41834982a88ac5e839c5561bf31cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d09a9432fcb41834982a88ac5e839c5561bf31cc"}]},"branch":"refs/heads/master"},"67854d125f15b743b27556f1eff6dbd9d34dbfe5":{"kind":"REWORK","_number":8,"created":"2025-12-17 12:47:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/62/969962/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/62/969962/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/62/969962/8"}}},"commit":{"parents":[{"commit":"28a372841402381245d65cca79c171074c4f5d21","subject":"Fix tap_mirror extension name in KNOWN_EXTENSIONS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/28a372841402381245d65cca79c171074c4f5d21"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2025-12-05 16:07:59.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-12-17 12:47:42.000000000","tz":0},"subject":"Add \"both\" direction to tap mirror definition","message":"Add \"both\" direction to tap mirror definition\n\nPreviously, only IN and OUT directions calls were supported. Now\nit also supports the BOTH direction.\n\nRelated-Bug: 2134096\nChange-Id: I6328c5977488066bcb8247111a7a629b88fd487a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/67854d125f15b743b27556f1eff6dbd9d34dbfe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/67854d125f15b743b27556f1eff6dbd9d34dbfe5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
