)]}'
{"id":"openstack%2Fneutron-lib~974248","triplet_id":"openstack%2Fneutron-lib~master~I6657269c1e70be74857913e66a455e5104f16f26","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-01-28 14:49:24.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-01-29 18:16:32.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I6657269c1e70be74857913e66a455e5104f16f26","subject":"Remove tenant_id from HANetworkConcurrentDeletion()","status":"MERGED","created":"2026-01-21 20:16:52.000000000","updated":"2026-02-03 20:39:20.000000000","submitted":"2026-02-03 20:38:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974248","meta_rev_id":"db7588bafdcee269ddad64ea11eb4bcf6b948f35","_number":974248,"virtual_id_number":974248,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-03 20:38:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2026-01-27 11:31:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-01-28 14:49:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2026-01-28 14:49:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-22 02:59:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-23 10:12:50.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-01-27 11:31:58.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"8521968993f502b5867c0ccf37cf31dcddc7b1d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-21 20:16:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c7e46d0dc40fb390c17f734ce6bb4df38b272da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-21 22:29:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c592f5d2f3e7f0e7238d3ed36994335abcc7bf04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-22 00:01:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b80fd945444355974f6a74db6f9d08abdb945d1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 02:59:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9534b05320454b079e99e33000fbb307\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91b139b6e8e44baabbc8e0646c505ec6 : SUCCESS in 6m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe4004a6b50b474a90fc0a3d046dc68a : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/277ced03bc594e4d90c4827c12d04cc4 : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e668e2886454fb391831cdfd468d2d2 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/429ed8b23ef64aa1a1dc306752b35c77 : SUCCESS in 6m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/401e7ae7709445aeae50b6c31d671b75 : SUCCESS in 1h 18m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/35ad0fd64147453ea5202b202a6599ac : FAILURE in 1h 19m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fa437b69532e4653a9bea160025bc9c8 : SUCCESS in 2h 51m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ab7998f6fce4306ba5ab4d95c2dff9e : SUCCESS in 29m 54s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/b0f2c56d38554c65adc9ce1e16ce483f : SUCCESS in 16m 21s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/de71405d45d64b3d81040c66c98582ea : SUCCESS in 8m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/956d63064a2c47f28b11e99965c672f6 : SUCCESS in 1h 45m 41s","accounts_in_message":[],"_revision_number":3},{"id":"a59d430e140aba5bbcd27ee89633d3660109a120","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-22 03:04:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f4dd312e1cad51e11805bda7fbcfcd0ca695763e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 04:56:25.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fcf0cc289e145db8fa2478ca5f2c356\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/921b9a06d7064046a57685df4aa6043b : SUCCESS in 8m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5312a650706d47a7af3fe46d97093dca : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fe0f9c47b634ba8a17ede200827e58b : SUCCESS in 6m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cf3c4523f5c49d0b3c02e82dcc022a8 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa5ca4e0906143cab8f59b35c9406669 : SUCCESS in 5m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5fdc6f65d16a4b4791bff5a5ff429e77 : SUCCESS in 52m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c864c58a339040f9aeb6f1e6684978f9 : SUCCESS in 1h 24m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f2dbff7488ba45a399bf0a7a1f76c87a : SUCCESS in 1h 51m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fad1a03f757c4e96bf726c3da92cd626 : SUCCESS in 32m 37s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/9e31e98f98624048b9f6727c908e68aa : SUCCESS in 29m 37s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6f6c08814e254b118a5b3707e76f2f96 : SUCCESS in 6m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/515ff802827f4414ad385e0e3fb4ec85 : FAILURE in 54m 38s","accounts_in_message":[],"_revision_number":3},{"id":"70e7c34bba8f54ad9ae03044638663b5fe6144df","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-23 04:24:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b8cd0aa1dffdb4b33421809328cae57efd7328a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 06:07:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc4c9028502544b8985715680f5a2e4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/798a9e1c2e964a09bb4913b7b054601c : SUCCESS in 8m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be758227647b47709a7a2607690b576d : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6598692144d941159d0cfce5838eb47c : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/408cb3a32dd34f41a3aa70d5945e3a99 : SUCCESS in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/040fd861a6bd4b81a4b57195436a2b09 : SUCCESS in 3m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3519407214140c8954d329c4992107f : SUCCESS in 55m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a78f5dcd0ca7471eaeedd90878e8d591 : SUCCESS in 1h 41m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/651c33967c97422ea91ec1a8e2826ca7 : SUCCESS in 1h 42m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d2d782bdeb914e7ba0342693781b6e7a : SUCCESS in 35m 45s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/b00978816cd64271845255816b4c263c : SUCCESS in 27m 31s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ce51c654c70a475ebf8e581dba53e34e : SUCCESS in 5m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ad372fb14d9648ab9a78e3718a4322e5 : SUCCESS in 1h 06m 24s","accounts_in_message":[],"_revision_number":3},{"id":"8016243786baca448a8cce5f41e183530a9a2d4a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-23 10:12:50.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9bb62e783755997e751aa576539d74e620e78670","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-25 01:15:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"80134fce6d5affd6e2e6a8ede7e5b005d1420d29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-25 01:16:46.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"30fc30f3292cfe0a75b703766ed9478b7915f6cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 03:54:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3672f049118b4a2488bf73edf0f74a49\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2691947a82f74586b6e5da5325d4865b : SUCCESS in 13m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3ab63801bb64b5aaa5cd337708b0b2b : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f193ae2334e432ba9e3029f16a5e423 : SUCCESS in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bb2cd8f604c4a51a2a9fc55452f711f : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ad81cbf4f1c4288af3a97227c88608f : SUCCESS in 2m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/662a070b0b324df9926d0b7c2aadb948 : SUCCESS in 1h 41m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1c3d423c623444d690d7ced12cde1941 : SUCCESS in 1h 39m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8e6ea7a9db0a4295b84be024b283e487 : SUCCESS in 2h 34m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6fe45cabf3f54a098cf5d7a2be0bc105 : SUCCESS in 31m 52s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/3c81feb044e3442194acb3751915e2bb : SUCCESS in 16m 21s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/64f15815503c483da568893bef9da66d : SUCCESS in 5m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3cf986aaaec7490598cbc6c71991d613 : SUCCESS in 1h 43m 51s","accounts_in_message":[],"_revision_number":4},{"id":"333c1f82c04bf6dae34bcf4246dbd327871fb6d1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-27 11:31:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"52662877406b99c6b64fd85f70fff7e39034046c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-28 14:49:24.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"537d3129d697a2e1cbd805388bc7e504c4150ed4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 14:49:30.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccad86048db547669e3e7c7a49fbaa1e\n\n\nWarning:\n  Change 974247,1 in project openstack/neutron does not share a change queue with 974248 in project openstack/neutron-lib","accounts_in_message":[],"_revision_number":4},{"id":"83ba72dc6ae0323129c016a341c160f965cdf451","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-29 18:16:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0f80c9c87c19c7880f20a0ecc6736f2f614858a8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-03 16:16:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7a797654a67f8fdd163d210ef555a2832877a4f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 18:24:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab959256024a4d38a4054bef69d6120e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e772b3ca84413eac202b4952b45a02 : SUCCESS in 8m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/213bd2548b29483a984555027a0654a8 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b835e28040d46d58236346dfad8564e : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6c0a966b929419183a8609fa7ee9f36 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d61ab8e9e9c4562a8473532bede5a0a : SUCCESS in 3m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78409def56b14cef88c76b9540d571a2 : SUCCESS in 1h 42m 13s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/091b1fa92aa14e7ab74c862eee281849 : SUCCESS in 1h 32m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bb9d79a09b7b419f848ab45757779535 : SUCCESS in 2h 06m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6a96a374544d4a09a09ae949c502eca8 : SUCCESS in 31m 33s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/af198f7fc4774124891567ebdb380cf2 : SUCCESS in 19m 07s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e286cf849a5344a7a9b4f2a7fed6febe : SUCCESS in 6m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ec0c5e8db14844158082e57c1977fc36 : SUCCESS in 1h 25m 04s","accounts_in_message":[],"_revision_number":4},{"id":"d5284630b26578f5433823ed51052e0d105992af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 18:24:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"32b80180c73b6a5fb11959a2093b05fd008dbe09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 20:38:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a53b58c847124d39a4d36896fcf50d25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/028bc60b9b754a869db764ee4462ec2a : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73cca5b343374c6181cfc8e3b58ee684 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/411416ef316840c28a5b102e2eba03d7 : SUCCESS in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4649a1fbcb1d4a85a00f2d0c5ffa625e : SUCCESS in 3m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/518a00f6588a4477a62840a2f15175bf : SUCCESS in 55m 10s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4d107817fbfd47a79f503e67089bb671 : SUCCESS in 1h 15m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f3d8f8cfc3344420ab1e87d81a887c9e : SUCCESS in 2h 12m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2338ddf7f4454a67a36bd83e94eb43c7 : SUCCESS in 45m 09s","accounts_in_message":[],"_revision_number":4},{"id":"e1b820f140fd0023d8b0cc738d6ab123df8b34c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 20:38:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"db7588bafdcee269ddad64ea11eb4bcf6b948f35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 20:39:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33e9f1efeb484c99a1f064202e1d2bd1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/984d3e9a5b42412b84ef71ecd3c06dec : SUCCESS in 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"283fa85e189dd22c5839a5b7c38f4cb313333374","revisions":{"f948ae7658b09919b6413d8be6f504430bc8ea0f":{"kind":"REWORK","_number":1,"created":"2026-01-21 20:16:52.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/48/974248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/48/974248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/48/974248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/48/974248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/48/974248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/48/974248/1"}}},"commit":{"parents":[{"commit":"5cf39169a498372f8019922f9c27277fe4cac8cc","subject":"Use py313 for all neutron-lib jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5cf39169a498372f8019922f9c27277fe4cac8cc"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-01-21 20:13:41.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-01-21 20:13:41.000000000","tz":-300},"subject":"Remove tenant_id from HANetworkConcurrentDeletion()","message":"Remove tenant_id from HANetworkConcurrentDeletion()\n\nNo callers pass tenant_id argument any more after\ndependency merges, remove support for it.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/974247\n\nChange-Id: I6657269c1e70be74857913e66a455e5104f16f26\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f948ae7658b09919b6413d8be6f504430bc8ea0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f948ae7658b09919b6413d8be6f504430bc8ea0f"}]},"branch":"refs/heads/master"},"8d579dfae4889a3187d4f9924f893d16fe59a7c6":{"kind":"REWORK","_number":2,"created":"2026-01-21 22:29:12.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/48/974248/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/48/974248/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/48/974248/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/48/974248/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/48/974248/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/48/974248/2"}}},"commit":{"parents":[{"commit":"5cf39169a498372f8019922f9c27277fe4cac8cc","subject":"Use py313 for all neutron-lib jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5cf39169a498372f8019922f9c27277fe4cac8cc"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-01-21 20:13:41.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-01-21 22:29:04.000000000","tz":-300},"subject":"Remove tenant_id from HANetworkConcurrentDeletion()","message":"Remove tenant_id from HANetworkConcurrentDeletion()\n\nNo callers pass tenant_id argument any more after\ndependency merges, remove support for it.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/974247\n\nChange-Id: I6657269c1e70be74857913e66a455e5104f16f26\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8d579dfae4889a3187d4f9924f893d16fe59a7c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8d579dfae4889a3187d4f9924f893d16fe59a7c6"}]},"branch":"refs/heads/master"},"a00cde225c3f2119059660b9de1cf8af0c38ba05":{"kind":"REWORK","_number":3,"created":"2026-01-22 00:01:34.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/48/974248/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/48/974248/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/48/974248/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/48/974248/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/48/974248/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/48/974248/3"}}},"commit":{"parents":[{"commit":"5cf39169a498372f8019922f9c27277fe4cac8cc","subject":"Use py313 for all neutron-lib jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5cf39169a498372f8019922f9c27277fe4cac8cc"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-01-21 20:13:41.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-01-22 00:01:28.000000000","tz":-300},"subject":"Remove tenant_id from HANetworkConcurrentDeletion()","message":"Remove tenant_id from HANetworkConcurrentDeletion()\n\nNo callers pass tenant_id argument any more after\ndependency merges, remove support for it.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/974247\n\nChange-Id: I6657269c1e70be74857913e66a455e5104f16f26\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a00cde225c3f2119059660b9de1cf8af0c38ba05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a00cde225c3f2119059660b9de1cf8af0c38ba05"}]},"branch":"refs/heads/master"},"283fa85e189dd22c5839a5b7c38f4cb313333374":{"kind":"REWORK","_number":4,"created":"2026-01-25 01:16:46.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/48/974248/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/48/974248/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/48/974248/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/48/974248/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/48/974248/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/48/974248/4"}}},"commit":{"parents":[{"commit":"55cc6fdaa9dcb1d4347d2b5762273624de46ac3e","subject":"Merge \"Rehome ovn-db-sync-util related constants to neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/55cc6fdaa9dcb1d4347d2b5762273624de46ac3e"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-01-21 20:13:41.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-01-25 01:14:37.000000000","tz":-300},"subject":"Remove tenant_id from HANetworkConcurrentDeletion()","message":"Remove tenant_id from HANetworkConcurrentDeletion()\n\nNo callers pass tenant_id argument any more after\ndependency merges, remove support for it.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/974247\n\nChange-Id: I6657269c1e70be74857913e66a455e5104f16f26\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/283fa85e189dd22c5839a5b7c38f4cb313333374"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/283fa85e189dd22c5839a5b7c38f4cb313333374"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
