)]}'
{"id":"openstack%2Fneutron-lib~976808","triplet_id":"openstack%2Fneutron-lib~master~Ia47fe33523c97495a1b6347d86510430509b2b34","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-02-17 14:36:15.000000000","reason":"Change was abandoned"},"36630":{"account":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"last_update":"2026-02-17 18:51:49.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-02-17 10:12:56.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}},"hashtags":[],"change_id":"Ia47fe33523c97495a1b6347d86510430509b2b34","subject":"Add the is_filter attribute to status on trunk port","status":"MERGED","created":"2026-02-13 19:06:20.000000000","updated":"2026-02-17 18:52:44.000000000","submitted":"2026-02-17 18:51:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976808","meta_rev_id":"ab7b5becc118b8654e10f4376df2daee07ea0e03","_number":976808,"virtual_id_number":976808,"owner":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-17 18:51:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-02-17 10:12:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-02-17 10:12:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-17 10:12:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-02-17 14:36:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-02-17 10:12:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 20:58:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-16 09:22:42.000000000","updated_by":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"reviewer":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"state":"REVIEWER"},{"updated":"2026-02-16 11:22:51.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-02-16 11:23:58.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-02-16 14:31:45.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-02-17 09:54:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"f507c817ce8f607790397c013b44714706923b37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"date":"2026-02-13 19:06:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6755d00f427417d5d82065fb40aa92d3ca2005b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 20:58:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8089e6816b5493cb01dfb76923d13f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c7a5b817e6d4b808309fca3312ea4e2 : SUCCESS in 8m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7099b0dcf4b04b1ba108f43dc81ca0ba : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be673e88bd3d4aac95af7acd8b2b1e3f : FAILURE in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53c1597bcd3f45aeabb4df289649ceae : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a938522feb344e81be3374fe1f3c21dd : SUCCESS in 5m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/098cc8c6ffb74c5ebfc0d409fd35af19 : SUCCESS in 52m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ac08f0f7e152402cb7387a92ba7ed1f3 : SUCCESS in 1h 25m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f1d0ab48228445c483074e99d7f16590 : SUCCESS in 1h 50m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ad9c93510995403385f0a6898223121c : SUCCESS in 34m 33s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/a62d9495446b4a8c9d56357eedcd31c4 : SUCCESS in 15m 38s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/92068ea5133b4f43ab9689a5c1826eec : SUCCESS in 5m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/69b5b2abf3c94e5d89221ae7603f7848 : SUCCESS in 1h 06m 15s","accounts_in_message":[],"_revision_number":1},{"id":"cd8d1849dee88203278432e58048de1b9fba51fb","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2026-02-16 09:22:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d346bce1978c11dfa2bc8133bea836340b6051d5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-16 11:22:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"468abbcb9d27701b089db48d1098713c5b32c0d6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-16 11:23:58.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"800f3aefad7ced65ea00ace339b832165f731c5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 13:42:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/adce068bd92f467683c2aa9a4c44a46a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/627df8d187ad4f5ba298800bf74ff006 : SUCCESS in 8m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94dc78eed15743739a59923097b670e7 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b6f73c8dc2e451ca3ee89841940ce43 : FAILURE in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/590f5d83214c4d349b6a11937a511182 : SUCCESS in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d82bbf74ff14bea90696accd621623a : SUCCESS in 5m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b001e798f964ec3bfe02145fee904ad : SUCCESS in 1h 39m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/400c3faf9fd649e68e8d2a8d5bc14b59 : SUCCESS in 1h 24m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/de2bad5c057a48f7ad0cceda80e630d9 : SUCCESS in 2h 18m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/90f2ec3aad544e21908f1ab5677e6880 : SUCCESS in 33m 16s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/14e3646e390442d3b1ab6a9f95720c7d : SUCCESS in 26m 28s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/516073f9c25e4730be955994ba6da5e4 : SUCCESS in 6m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d1659203c5b64cf2b3fb18859499f90b : SUCCESS in 1h 16m 28s","accounts_in_message":[],"_revision_number":1},{"id":"64eb6066a4ea5a86ce45ea30ca2421fe7e61904f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-02-16 14:31:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93237c856a7eecfb42098b89dffea3c9c084ef6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 16:26:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/412184e570df48b8aa548f8122341ade\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08fa092ea5cd4c66b3f903d1855f0244 : SUCCESS in 8m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/532a968ea0d2465f8c55274f8632c33d : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a87b5e7fa9344f4d9a3e0257b44c49e9 : FAILURE in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b1cde2f213e452c883be0628f61dd71 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/426b6305d23546a5947abce73afc5876 : SUCCESS in 3m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef63c203983748259e3c7895d30ed98a : SUCCESS in 1h 37m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f03044fe68894866b60b52c913d808e3 : SUCCESS in 1h 26m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b5513716918b48c596d45db3037ea95c : SUCCESS in 1h 51m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/366a6a3b1b1246ca83f38c190eb09274 : SUCCESS in 35m 43s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/7e5b28e4f201435db4c0343ca0da15b4 : SUCCESS in 26m 20s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7852673c8c8a496290afa593c574fbe9 : SUCCESS in 6m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5a1149e3532c4a32b25dcd16c4110aee : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":1},{"id":"34c0a24c7f4d3bbb23fd618696c07785b7b2c540","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"date":"2026-02-16 22:47:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67752af412f4a05f67a22d03b4cd347b19a4606d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"real_author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-02-17 09:52:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_36630\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"}],"_revision_number":2},{"id":"24210f9f8fdb8b67a9218736a40e9076d820c647","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-17 09:54:50.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4f98d9620ec51dbd5dfb96c91052eda292323292","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"real_author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-02-17 10:12:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_36630\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1\n","accounts_in_message":[{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"}],"_revision_number":3},{"id":"c591c86c72a7a68b90c2cd64bb96ab76519b660f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-02-17 10:12:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8776961c973c8517d2084225bf1b6cf09e6c28a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 12:38:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d744356f07124df08347e0f76eca6a16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dbfcd2b73b34fce961d97c6744d981c : SUCCESS in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25b0a18830834647b085f74f4967b77e : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/921c121103af4b46a65662a7018f53a5 : SUCCESS in 7m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b895823464884789a85e91fe9d7fd9fe : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/54bbd048371046f5bb906f6c2d80c57c : SUCCESS in 2m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51260463aa3642a6b1e961e5f1a02d8e : SUCCESS in 59m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ccd4c45c0e694d758f53360a4b5e151f : SUCCESS in 1h 41m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f8933b48b30645ea86a51f6f19a9c957 : SUCCESS in 2h 22m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/754de4e5b27443468070120e2fe40d2e : SUCCESS in 32m 07s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/27591bf0ab2d4fe5a837f01e154c8e73 : SUCCESS in 16m 09s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5f4aa7bcc5d54bbf90842fbf3a8c02e0 : SUCCESS in 7m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/95b90baf61fc493f9323cc45218c4204 : SUCCESS in 1h 19m 03s","accounts_in_message":[],"_revision_number":3},{"id":"73134ce86bff4a0cee1fb99a0d6ac98b46d6ab21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 12:38:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ab80208323152a74fddb563ed650fe70c5459121","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-17 14:36:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3},{"id":"cbfb715dcbb0bb58e591f91abcf37ef3dbecbb03","tag":"autogenerated:gerrit:restore","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-17 14:36:22.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"068779c4d64705a7c153e288b047435c3c394518","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-17 14:36:31.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6f8ae847fbaae72a1ca177f0a2241dfa3e8c5f3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 16:37:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3723a742c6db4b85b8768db60f6168a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/543e81e889dc4713bc96ab52f1d666f2 : SUCCESS in 9m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9facf52a3d34a8f885615ac7affea66 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19ab5e0f52c24f439fcc22e710bbdb43 : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e022e93db4d34f958e91dbbf8cd0014a : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27c144c5c9d0463096d3169c58194bf6 : SUCCESS in 4m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad04ed187ece445f9194b10e7260548c : SUCCESS in 1h 48m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/33ba3726b3284e4099c60949002bca5a : SUCCESS in 1h 21m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b4f1c3e551ca40c8a639c686e731c5d4 : SUCCESS in 1h 57m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f8d35534477a4ae4a8b8fa13464df099 : SUCCESS in 30m 55s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/9865ad3a361f461f8f709f77811cb929 : SUCCESS in 20m 08s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/afa24f555d304c228e6a64d44410cadd : SUCCESS in 6m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/855bcbcd8de8403f9912410f7d2f4045 : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":3},{"id":"a10dca26976312ff3c9424d3ff359385d2e72219","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 16:37:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"eb63c5eda5d916c3c59815a63b2e8b154376f789","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:51:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c3babafc54a41a6b6f6441326110d9b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/958403495b514fbbabdad299edfa638d : SUCCESS in 11m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c920a3b8621d4a23be5027d5aa2016ee : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae22b0e439f14975924aa2acdb4580be : SUCCESS in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f3bad6e1bea84c3faf96d61657c53289 : SUCCESS in 3m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e17b7ede6ec49498215313e917b0aa9 : SUCCESS in 2h 00m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fc73ba92eb7946a0824eff74866d332a : SUCCESS in 1h 26m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d1e398e5ffb34053b0be463b54aa4f7e : SUCCESS in 1h 56m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/89b9b3f35be04a54800e6118f881056f : SUCCESS in 1h 14m 29s","accounts_in_message":[],"_revision_number":3},{"id":"1241bfda91a8a30fe30a73dfc6909368b0e1603b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:51:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ab7b5becc118b8654e10f4376df2daee07ea0e03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:52:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22915a1e38c4499d8aa7fb0c5ef7d5a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb87b7d657d6499f83539515e8d401dd : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b1ee750c932da1b0027b0b653ea6a566742ee2b9","revisions":{"baa43f9368cfa8439e542ff21079ab48e0afc17c":{"kind":"REWORK","_number":1,"created":"2026-02-13 19:06:20.000000000","uploader":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"ref":"refs/changes/08/976808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/08/976808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/976808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/976808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/976808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/08/976808/1"}}},"commit":{"parents":[{"commit":"139b452617302ae15e6997fcc9d89f475a54e13a","subject":"Merge \"Remove tenant_id from HANetworkConcurrentDeletion()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/139b452617302ae15e6997fcc9d89f475a54e13a"}]}],"author":{"name":"Winicius Silva","email":"winiciusab12@gmail.com","date":"2026-02-13 18:59:23.000000000","tz":0},"committer":{"name":"Winicius Silva","email":"winiciusab12@gmail.com","date":"2026-02-13 19:04:45.000000000","tz":0},"subject":"Add the is_filter attribute to status on trunk port","message":"Add the is_filter attribute to status on trunk port\n\nWe are not able to filter trunk port by its status since it is failing\non validating the available filter. So this patch adds the is_filter to\nthe status field.\n\nCloses-Bug: #2141784\n\nChange-Id: Ia47fe33523c97495a1b6347d86510430509b2b34\nSigned-off-by: Winicius Silva \u003cwiniciusab12@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/baa43f9368cfa8439e542ff21079ab48e0afc17c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/baa43f9368cfa8439e542ff21079ab48e0afc17c"}]},"branch":"refs/heads/master"},"e9481fdf2b1939789a3142846662e50e1b95bd89":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-17 09:52:32.000000000","uploader":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"real_uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/08/976808/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/08/976808/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/976808/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/976808/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/976808/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/08/976808/2"}}},"commit":{"parents":[{"commit":"8664e618c9a0192d706886bf7b09adbb465c371f","subject":"Fix issues related to ``pkg_resources`` module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8664e618c9a0192d706886bf7b09adbb465c371f"}]}],"author":{"name":"Winicius Silva","email":"winiciusab12@gmail.com","date":"2026-02-13 18:59:23.000000000","tz":0},"committer":{"name":"Winicius Allan Bezerra da Silva","email":"winiciusab12@gmail.com","date":"2026-02-17 09:52:32.000000000","tz":0},"subject":"Add the is_filter attribute to status on trunk port","message":"Add the is_filter attribute to status on trunk port\n\nWe are not able to filter trunk port by its status since it is failing\non validating the available filter. So this patch adds the is_filter to\nthe status field.\n\nCloses-Bug: #2141784\n\nChange-Id: Ia47fe33523c97495a1b6347d86510430509b2b34\nSigned-off-by: Winicius Silva \u003cwiniciusab12@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e9481fdf2b1939789a3142846662e50e1b95bd89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e9481fdf2b1939789a3142846662e50e1b95bd89"}]},"branch":"refs/heads/master","description":"Rebase"},"b1ee750c932da1b0027b0b653ea6a566742ee2b9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-02-17 10:12:47.000000000","uploader":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"real_uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/08/976808/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/08/976808/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/976808/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/976808/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/976808/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/08/976808/3"}}},"commit":{"parents":[{"commit":"feebd9b011ce1d3d32e10fa6d93083b897c63e8e","subject":"Fix issues related to ``pkg_resources`` module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/feebd9b011ce1d3d32e10fa6d93083b897c63e8e"}]}],"author":{"name":"Winicius Silva","email":"winiciusab12@gmail.com","date":"2026-02-13 18:59:23.000000000","tz":0},"committer":{"name":"Winicius Allan Bezerra da Silva","email":"winiciusab12@gmail.com","date":"2026-02-17 10:12:47.000000000","tz":0},"subject":"Add the is_filter attribute to status on trunk port","message":"Add the is_filter attribute to status on trunk port\n\nWe are not able to filter trunk port by its status since it is failing\non validating the available filter. So this patch adds the is_filter to\nthe status field.\n\nCloses-Bug: #2141784\n\nChange-Id: Ia47fe33523c97495a1b6347d86510430509b2b34\nSigned-off-by: Winicius Silva \u003cwiniciusab12@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b1ee750c932da1b0027b0b653ea6a566742ee2b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b1ee750c932da1b0027b0b653ea6a566742ee2b9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
