)]}'
{"id":"openstack%2Fneutron-lib~977313","triplet_id":"openstack%2Fneutron-lib~stable%2F2025.1~Ia47fe33523c97495a1b6347d86510430509b2b34","project":"openstack/neutron-lib","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"36630":{"account":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"last_update":"2026-03-21 01:27:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia47fe33523c97495a1b6347d86510430509b2b34","subject":"Add the is_filter attribute to status on trunk port","status":"MERGED","created":"2026-02-19 11:42:31.000000000","updated":"2026-03-21 01:28:41.000000000","submitted":"2026-03-21 01:27:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977313","cherry_pick_of_change":976808,"cherry_pick_of_patch_set":3,"meta_rev_id":"f026cd4ca783bdeb8523e919911b8228c68d7c11","_number":977313,"virtual_id_number":977313,"owner":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-21 01:27:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-03-03 14:01:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-02-21 22:37:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-03-03 14:01:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-19 13:40:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-21 22:37:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-03-03 14:01:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"ff460aff1f8ff5ecde5053c8a02dd15b869d2bdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"date":"2026-02-19 11:42:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f922963628482de32be8a37708993e5cc05ead8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 13:40:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb389d435e4241a19cb30dc6dde2430c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a99c19eff7714137af1a20295b0b864e : FAILURE in 5m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be0bbf618bd14882a16fc5dac83d4013 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ac5621f169c4440be9714d2a5fcb5af : FAILURE in 2m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa89effee83d41c89fb6467f7c09f2f0 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b02578807d064b878cf6e30de580a3b0 : SUCCESS in 6m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8a5decec5c34d6383a6aa89ce557520 : SUCCESS in 1h 00m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/32580f589d8d43cc838e3c63886a7134 : SUCCESS in 1h 29m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ba2722029c3f452585cd6cea35735dc1 : SUCCESS in 1h 56m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/396457d237474b7ca3144dff1e81678d : SUCCESS in 32m 12s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/67ed1cbf6895478eb88712e348467859 : SUCCESS in 22m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/824a59e788e347ae850e78884bbfc52d : SUCCESS in 1h 17m 20s","accounts_in_message":[],"_revision_number":1},{"id":"5f363146498dbd97f5e9ff8d438fdc6a1bf6e68e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-21 22:37:27.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5312ec1f0c32681bcbdb16d5d2db4effc1302c8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 00:21:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c4ae376b83a436d94c42fa2346beb9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aab7aef148e44b69c8a72cdd1105deb : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb4e653563534bc79a35c1a70231b2ec : SUCCESS in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66d29086456a495382ea1fe521b5df76 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bc0fc3cfb4743d5bdf6ecbab781aa0f : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4e8cb2b21fe4a358b7a3843c04da705 : SUCCESS in 2m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66c38ab4375b40c2a26fc2c886428f4e : SUCCESS in 58m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5aa2cfb7f6114770a3cd196480a11d48 : SUCCESS in 1h 29m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/88d5323c6289486cb84f6e736ff65614 : SUCCESS in 1h 43m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c5855b00402b4c299c08259dec719fbb : SUCCESS in 32m 42s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/617d1851dc69413ea57d5fa53f026b60 : SUCCESS in 16m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/36d85a850b2a45148ddc531301cdad00 : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":1},{"id":"467ff93d9cba9fe8612d9b03865f66fa82e47be6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-03 14:01:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"acfee07a8a389743806f24a9c10ee72b3e2ca893","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 14:02:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71331750d30559dc02f92484aa71a99342953995","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 16:10:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbe4790d05ad4975af105dca14c618e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71cdd224f01b4e72beb11066dfe777ec : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9b1b32d35c6453c9974796b79416d6e : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22a88e624c0741f193fb2522d3bf2d44 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/265635fcfe4d4e2b93f368523ed42a84 : SUCCESS in 2m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d7de4a5b1d04851ba60d0711839cccd : FAILURE in 51m 06s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9f5b20b7fbac48178f8702131f25961c : SUCCESS in 1h 32m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/21a8e8ab7977463ba4876473d43e051e : SUCCESS in 2h 07m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/539e15156d6c4389bb4f35eb2dd805a9 : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":1},{"id":"de0811b672dbaaf4c983431ce5883ce11a473b10","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-20 21:21:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b79be0023975ef349915c9611040cde4fe45914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 23:18:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02ad783cd7f8464481c13292a25192c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80c5d7fd060f47c1a67cb3577d82aa3f : SUCCESS in 10m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04f64cd7e13a4ba6a8f31f48f8fcd3fb : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3750d4ce0a2a4852a56bf4b108fd87bf : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b547e15db8bd479685d8c384003cd03a : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/920766e1029142608f66cf4c811f53cd : SUCCESS in 3m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3154c982ab8344d38f1646bbc3051de5 : SUCCESS in 58m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/72354813485647a7860d6f1c0ace0514 : SUCCESS in 1h 14m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8b42432dbdf44f818da0bc23548cc7e4 : SUCCESS in 1h 56m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/60c04bd14cea45d29e35d309e46440f3 : SUCCESS in 33m 02s\n- openstack-tox-py312-with-neutron https://zuul.opendev.org/t/openstack/build/1b5909189aae431c942664818dfa1fa5 : SUCCESS in 23m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/13f3ec699ac5450fab3be61e51a850ab : SUCCESS in 1h 05m 31s","accounts_in_message":[],"_revision_number":1},{"id":"81f5cbd4cfe07b44771ee32afc54ec09b8c101c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 23:19:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"302328afe37d1e35e41b20146b834e11cdc2c2ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 01:27:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7696f47567a74ef8bf57b6f92550407a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04125e849d294d86bed28192173b16f7 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6923249e1477458c86d6eb3aad0cbe91 : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef8da360f09a4fc79bdb8cd40754c7f5 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3225b28bba314fd186b4baa3a4062b04 : SUCCESS in 4m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c49c39a4c75c48b391011fb58829b27d : SUCCESS in 53m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bac3aab11639481e8c8be4f0a5c23718 : SUCCESS in 1h 35m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8839c9bd9e4946d7b1fad80295f2dc2c : SUCCESS in 2h 08m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/390999c1df9744a39e8b308a4afaf14f : SUCCESS in 1h 01m 42s","accounts_in_message":[],"_revision_number":1},{"id":"2f1823cb65a69f1476439647078cc5f93298c971","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 01:27:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f026cd4ca783bdeb8523e919911b8228c68d7c11","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 01:28:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b6db510ef104596b1465159258253b8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4965c57a11f94f05abb0a9f7407fdf1c : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ad016f0f37e9d98f1e6d394d45bc1bfea763fc4","revisions":{"9ad016f0f37e9d98f1e6d394d45bc1bfea763fc4":{"kind":"REWORK","_number":1,"created":"2026-02-19 11:42:31.000000000","uploader":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"ref":"refs/changes/13/977313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/13/977313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/13/977313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/13/977313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/13/977313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/13/977313/1"}}},"commit":{"parents":[{"commit":"080219df171caee02fa85e4c67a66fe103e6a6a8","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/080219df171caee02fa85e4c67a66fe103e6a6a8"}]}],"author":{"name":"Winicius Silva","email":"winiciusab12@gmail.com","date":"2026-02-13 18:59:23.000000000","tz":0},"committer":{"name":"Winicius Allan Bezerra da Silva","email":"winiciusab12@gmail.com","date":"2026-02-19 11:42:31.000000000","tz":0},"subject":"Add the is_filter attribute to status on trunk port","message":"Add the is_filter attribute to status on trunk port\n\nWe are not able to filter trunk port by its status since it is failing\non validating the available filter. So this patch adds the is_filter to\nthe status field.\n\nCloses-Bug: #2141784\n\nChange-Id: Ia47fe33523c97495a1b6347d86510430509b2b34\nSigned-off-by: Winicius Silva \u003cwiniciusab12@gmail.com\u003e\n(cherry picked from commit b1ee750c932da1b0027b0b653ea6a566742ee2b9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9ad016f0f37e9d98f1e6d394d45bc1bfea763fc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9ad016f0f37e9d98f1e6d394d45bc1bfea763fc4"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
