)]}'
{"id":"openstack%2Fneutron-lib~977314","triplet_id":"openstack%2Fneutron-lib~stable%2F2024.2~Ia47fe33523c97495a1b6347d86510430509b2b34","project":"openstack/neutron-lib","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"36630":{"account":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"last_update":"2026-03-05 23:24:15.000000000","reason":"\u003cGERRIT_ACCOUNT_36630\u003e replied on the change","reason_account":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"}}},"hashtags":[],"change_id":"Ia47fe33523c97495a1b6347d86510430509b2b34","subject":"Add the is_filter attribute to status on trunk port","status":"ABANDONED","created":"2026-02-19 11:42:47.000000000","updated":"2026-04-07 08:23:52.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":976808,"cherry_pick_of_patch_set":3,"meta_rev_id":"1bd6ee1c7ae5dc76a92fbeb5d9efb6bc494958eb","_number":977314,"virtual_id_number":977314,"owner":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-19 13:31:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-19 13:31:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-03 14:02:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-04-07 08:23:52.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"a8be4c5c97a886586f141a154d0e93e3d50701fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"date":"2026-02-19 11:42:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ea63d53a6bf6625bc00408b6330dcffb05dc918e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 13:31:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47b11bc8c51b4c5c9b4a182c01840ed2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c7d4c6de1be4af78104a1bcfd86386a : FAILURE in 3m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c560d9f23c7459b9650ececf03b4dc0 : POST_FAILURE in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f221420763c42e586a44494118d2428 : FAILURE in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49d11a0252c74cd4a890280be3df838b : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/802013e4e76441d0a62b95514e1a5b9b : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d0d43614d574be1a4e276c50b31af25 : SUCCESS in 4m 27s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8db2671e1cb4e31be01a768da40ac06 : POST_FAILURE in 1h 33m 31s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/267986c106aa4e3a888c90c41da1cf23 : POST_FAILURE in 1h 12m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/3c63957604a9477cbac2307bd708301a : POST_FAILURE in 1h 33m 18s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/93b0ce3977254e5db79af5bf404cb077 : POST_FAILURE in 1h 44m 22s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/dc3e95d8cb684958a492b072c38504e4 : POST_FAILURE in 31m 36s\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/1667b23e214c4ca886dc31c21317e8f9 : FAILURE in 2m 34s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d0aead14471149b3a41d0e403dfaf8e6 : FAILURE in 23m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0cb2ef404af60b8691fc21f0755f2f5b24f9e7f6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-03 14:02:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"454c933b00184663f43eee4b2e4c3d1458f86a4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 15:59:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38e12939777944b59cfcc6b7ffb9f676\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d1d2cf931f4cd9a98449fa5f092950 : FAILURE in 4m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31492606234b465ebed2d60ad3eaee9b : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9694ec33c4904ee6a9ed5d6ad3e9215c : FAILURE in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4aee13d42ace4de095d683d09e6d75ce : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf9aca7a6cf345d48d7a796a041a465f : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/612d05b4c6854d52a451ddcb12419465 : SUCCESS in 5m 56s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8cea28335cf445c871c2891c5ab9585 : SUCCESS in 1h 12m 31s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/5860ad27bc6e43ddacec925ad395d2e2 : SUCCESS in 1h 26m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/9ea052d0ce3f4f9db051c20f48e37586 : SUCCESS in 1h 55m 41s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/e385938f83204c269548d609c1494b71 : SUCCESS in 1h 46m 26s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/88453640548244b39551e84fd2a5b53b : SUCCESS in 32m 36s\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/e8d93bb27edd4cb3a1ae3c5d98dbbd2e : SUCCESS in 20m 53s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a1b1240d5ac346838d8e231428558ea7 : SUCCESS in 1h 34m 53s","accounts_in_message":[],"_revision_number":1},{"id":"26d8b6642a26dfe1c3707bb53867073434c9738a","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"date":"2026-03-05 23:24:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1bd6ee1c7ae5dc76a92fbeb5d9efb6bc494958eb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-04-07 08:23:52.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e2a3fd1c9fa14a4068f4c847e254652da340666","revisions":{"9e2a3fd1c9fa14a4068f4c847e254652da340666":{"kind":"REWORK","_number":1,"created":"2026-02-19 11:42:47.000000000","uploader":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"ref":"refs/changes/14/977314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/14/977314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/14/977314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/14/977314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/14/977314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/14/977314/1"}}},"commit":{"parents":[{"commit":"1f6b35a0f39c9f594439084fd846a73fc7c48e2f","subject":"[2024.2 only] Switch to branched neutron-tempest-plugin jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1f6b35a0f39c9f594439084fd846a73fc7c48e2f"}]}],"author":{"name":"Winicius Silva","email":"winiciusab12@gmail.com","date":"2026-02-13 18:59:23.000000000","tz":0},"committer":{"name":"Winicius Allan Bezerra da Silva","email":"winiciusab12@gmail.com","date":"2026-02-19 11:42:47.000000000","tz":0},"subject":"Add the is_filter attribute to status on trunk port","message":"Add the is_filter attribute to status on trunk port\n\nWe are not able to filter trunk port by its status since it is failing\non validating the available filter. So this patch adds the is_filter to\nthe status field.\n\nCloses-Bug: #2141784\n\nChange-Id: Ia47fe33523c97495a1b6347d86510430509b2b34\nSigned-off-by: Winicius Silva \u003cwiniciusab12@gmail.com\u003e\n(cherry picked from commit b1ee750c932da1b0027b0b653ea6a566742ee2b9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e2a3fd1c9fa14a4068f4c847e254652da340666"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e2a3fd1c9fa14a4068f4c847e254652da340666"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
