)]}'
{"id":"openstack%2Fneutron-lib~977385","triplet_id":"openstack%2Fneutron-lib~stable%2F2024.2~I36cdabe3cfcc89ed9f2da7e1ae8d70fa77061e60","project":"openstack/neutron-lib","branch":"stable/2024.2","topic":"pkg_resources-stable/2025.1-stable/2024.2","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-04-08 11:23:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I36cdabe3cfcc89ed9f2da7e1ae8d70fa77061e60","subject":"Fix issues related to ``pkg_resources`` module","status":"MERGED","created":"2026-02-20 00:09:53.000000000","updated":"2026-04-08 11:24:28.000000000","submitted":"2026-04-08 11:23:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977385-pkg_resources-stable/2025.1-stable/2024.2","cherry_pick_of_change":977378,"cherry_pick_of_patch_set":3,"meta_rev_id":"e83edd228e9c2f40e9798a5a8c259fe7c4e8f3c7","_number":977385,"virtual_id_number":977385,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-08 11:23:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-04-06 14:27:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-04-08 09:38:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-04-06 14:27:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-20 00:09:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-02-20 01:23:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-06 14:27:25.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"dcd96b59f9fde3ac2e98fdfd8e2f937d95d327c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-20 00:09:53.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.","accounts_in_message":[],"_revision_number":1},{"id":"4446f5575f141112fcf006f45e159f3e73293c77","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-20 00:14:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d2d2ea8ae142d56a96204a847366de08f2bbbf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 01:23:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b86319daa1749f79a4c91b12cc1ffd3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b291c126a96c433ba38b9940ed02f4f1 : SUCCESS in 1m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d21783a7ade74449a60a13aada3fa7e4 : SUCCESS in 6m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47f51be075d3470e9ce8ef74142e3e45 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f017fa4974cb464eb2a0a8a85d382181 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f43dd95bc36456f83def82afea5ba99 : SUCCESS in 6m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f5e8b31c32c42bcaea0f9edc2808ebf : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/87b63105c0f1493090d61cf5f84cc4d4 : SUCCESS in 3m 55s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f04e849f23d1487dabbf42b7ecce380f : SUCCESS in 1h 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ddbafc467194ebe95e90f1643d542e3 : SUCCESS in 2m 12s\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/56c2f8b139d440709f1d5e51254c3aa3 : FAILURE in 4m 07s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/19fde500c20f4634907a68fe3ef3d4bb : FAILURE in 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e07f7a1b842406e7373e7b3b43b8d58db83219ca","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-20 15:23:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b6efec1b500efc988c4ac96a9cbba020f6b0ccb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-01 23:47:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3beeb934afd067f73fddc9313128559073b4db1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 01:27:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f0b366504bf4b8db6e6f35248163c26\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b93f40f4799341ad96d109bbac8a5e54 : SUCCESS in 2m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd9d251e300c4688b07d90faacdbd782 : SUCCESS in 4m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14bda795d01c425d8eaf3709b2abc306 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c81c55b22594d98ac65863539c71c62 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57863248f8c444fa82385a4404df5ded : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/39ffb9be381c4b07a1c17da4d4226fc9 : SUCCESS in 2m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/147e73a546f44afdb091360115a6b634 : SUCCESS in 4m 29s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a4e15c803eb4f678561f059f23f9a4a : SUCCESS in 56m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0be4a851feb34dbdae7cc43167b0a299 : SUCCESS in 3m 45s\n- openstack-tox-py311-with-neutron https://zuul.opendev.org/t/openstack/build/a98dcbb331ab43fab0a372effc9c8027 : SUCCESS in 21m 42s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c8d7109917a4dc8a3c0bf3797d22aa6 : SUCCESS in 1h 32m 33s","accounts_in_message":[],"_revision_number":1},{"id":"5ba0c635572dc46d2edb2fd07026405fb4d62697","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-06 14:27:25.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"039401d7decf3b259063f98203415222fd555183","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-08 09:38:47.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d80601566aabfc9ac81f063ebc4d81afae364ed0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 09:39:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73f552b039e29f21da7f2675eaf1d445e1087d92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 11:23:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7929ad7efed149ecbda538058aed52be\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/87a5d89b038c4d4bae95946b69d314d7 : SUCCESS in 1m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d59519567644ad79f4403e7380018ed : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba00740e5fa04aa69b44c726872bfc68 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/041ef21390f543cb92f5f504e8ba9b15 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a0e8a9a4c3e48d48d60e8df8d9ad5c2 : SUCCESS in 4m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9529c98a4d7420e9d6350d733060109 : SUCCESS in 1h 38m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0492d33955cb4e3088bc43adf28f7556 : SUCCESS in 4m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c2e0695471f46bb90249dfce2d8be9e : SUCCESS in 1h 19m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8f041f84d1b85fe2a57262b4529f280986992b03","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 11:23:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e83edd228e9c2f40e9798a5a8c259fe7c4e8f3c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 11:24:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8bb40b0e4cc4c039b4533bc158ce74d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b31b7a075d04af5a464cfb0aac8feaa : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa497edc801746b3b91d666225102b2d : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f822a0abbac4f0ddf0bd6e46c8a7a298afb9e813","revisions":{"f822a0abbac4f0ddf0bd6e46c8a7a298afb9e813":{"kind":"REWORK","_number":1,"created":"2026-02-20 00:09:53.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/85/977385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/85/977385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/977385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/977385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/977385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/85/977385/1"}}},"commit":{"parents":[{"commit":"1f6b35a0f39c9f594439084fd846a73fc7c48e2f","subject":"[2024.2 only] Switch to branched neutron-tempest-plugin jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1f6b35a0f39c9f594439084fd846a73fc7c48e2f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-16 10:16:41.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-20 00:09:53.000000000","tz":0},"subject":"Fix issues related to ``pkg_resources`` module","message":"Fix issues related to ``pkg_resources`` module\n\nThe library ``pkg_resources`` is deprecated and removed from\n``setuptools`` 82.0.0. This patch bumps ``flake8-import-order``\nto 0.19.0.\n\nFor stable/2025.1 and earlier this is a bump in\ntest-requirements.txt and .pre-commit-config.yaml does not\nhave the hacking check. Also had to pin setuptools in\ntox.ini.\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I36cdabe3cfcc89ed9f2da7e1ae8d70fa77061e60\n(cherry picked from commit feebd9b011ce1d3d32e10fa6d93083b897c63e8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f822a0abbac4f0ddf0bd6e46c8a7a298afb9e813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f822a0abbac4f0ddf0bd6e46c8a7a298afb9e813"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
