)]}'
{"id":"openstack%2Fneutron-lib~979320","triplet_id":"openstack%2Fneutron-lib~master~I4e0185af5173ecec39b010eed063a95f8ef84380","project":"openstack/neutron-lib","branch":"master","topic":"missing_string_formatter","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-03-09 14:34:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4e0185af5173ecec39b010eed063a95f8ef84380","subject":"Add missing string formatter","status":"MERGED","created":"2026-03-06 17:01:07.000000000","updated":"2026-03-09 14:36:42.000000000","submitted":"2026-03-09 14:34:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979320-missing_string_formatter","meta_rev_id":"2294c2d114479ae445186ab19329d2200fae90ad","_number":979320,"virtual_id_number":979320,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-09 14:34:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-03-09 12:13:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-03-07 00:01:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2026-03-09 12:13:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-06 18:53:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-07 00:01:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-03-09 12:13:30.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"82fb352c6c3f3baac3e2781d096df143afe64226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-06 17:01:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07be81c6f0942336e52e2c87e2a11a7e7ec09b17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 18:53:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e41acb3eed34015aab9d2b4d7434060\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c893a66655b4a33bdcd15510b0827ee : SUCCESS in 8m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/438c9c66d30e4b20bef7887684b601d7 : FAILURE in 1m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05fcc9c0f9f74357ba982a1dc099e66c : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/939a99b00db04906b5a22bd32dce3747 : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7fd7cf1db1124a9c97c4e0211968ace5 : SUCCESS in 3m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4f39bcdb10f940b09978cbffc3ceece0 : SUCCESS in 8m 23s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20f63d1c7846443f88607ab5226f59fa : SUCCESS in 1h 42m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f016ab13bd2945048d3c92077c573775 : SUCCESS in 1h 26m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bd8324e88054493cbe096a013b4e15a2 : SUCCESS in 1h 51m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/af6b5dcd9f774253b7e8a867bd973bcd : SUCCESS in 31m 30s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/8a13e8e1a28d42cab211b68a62ba02e9 : SUCCESS in 24m 37s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5b089c492e1541fe8c4f953f9be39ed7 : SUCCESS in 4m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9c7605d1838c4a9892b861ffdb81b110 : FAILURE in 20m 00s","accounts_in_message":[],"_revision_number":1},{"id":"afeebeea1f51880878ff9be35dee33b8067bb805","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-07 00:01:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cf2c48a08d9c26dbd16221ce9879798f3c49c4af","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-09 06:52:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"776e52b1dc78ccdc2c71f1af8c4d78ac552f4f2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 08:59:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b039cfa56314ad589ffa4b7d7a1043f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/033c692d1591451e96439338e4510771 : SUCCESS in 6m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61e1600fad0042dda7bdb3a6fe731a25 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae14f20c8bc456297652a75ed690fb6 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/450514b74a72418bb65ae5c93c5175a1 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb0dcb01692c43939380f8739fa83ed5 : SUCCESS in 3m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9e3c4201350949dfbd0adae2efef494d : SUCCESS in 5m 29s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/726e0482521d474184835d85a3774df6 : SUCCESS in 1h 35m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/aa75e56c2f7d4d2c903ecd6c65c32b53 : SUCCESS in 1h 28m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2685a102248d4192af95f76c0977ee91 : SUCCESS in 2h 05m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9e6956e4ad3b46b5a968566a77d109c5 : SUCCESS in 30m 41s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/c2159f5a307c46aa9e8883cb2835b5a7 : SUCCESS in 18m 29s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/86f83b556363460c8a8aa0333364043b : SUCCESS in 3m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cb07ff7005f548b1953a94bd8d236133 : SUCCESS in 1h 19m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8511da89fb2fe4c0d2dabc232ccb084d4dc02550","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-09 12:13:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9da85e9e268c0c2a030e250331231c5395efed1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 12:13:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe3ff5514db5a0e2bee11b7d400e7d4bb8d5cc27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 14:34:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/992e2c9bce9b4a2589d0aa71fa8c8829\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/974b9b09c7be4bb48122dcf39fee7372 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4143229b9506436fb326923fdfa7536e : SUCCESS in 8m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2cb84ea45bc43e880ce2e4ac76a72e7 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b3bbdf452d514ded8be5de862ccba17e : SUCCESS in 3m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5be38d3468b040a79cd9e34694540c0c : SUCCESS in 1h 51m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/87920962353e4b579f9d832dc630cdfd : SUCCESS in 1h 31m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/47f50ee685b1487c8f61f4fa72aef856 : SUCCESS in 2h 19m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1fd557b3f1c349afa160febf4fff01ad : SUCCESS in 1h 18m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b97932abfc7d423fa126a2a5e4f302267e014535","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 14:34:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2294c2d114479ae445186ab19329d2200fae90ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 14:36:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a50ab7f58b7e47f0ada95e40bd4c6b14\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2db1bbd1a15d4102a71801cb9c440ecc : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ead4a36410c4e9558707a37f84128f4783568b38","revisions":{"ead4a36410c4e9558707a37f84128f4783568b38":{"kind":"REWORK","_number":1,"created":"2026-03-06 17:01:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/20/979320/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/979320/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/979320/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/979320/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/979320/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/979320/1"}}},"commit":{"parents":[{"commit":"2cc6c9dbd29662d7a57ce3fd71016f5b735f5e93","subject":"Merge \"Add VIF_TYPE_VIRTUAL constant to portbindings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2cc6c9dbd29662d7a57ce3fd71016f5b735f5e93"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-06 16:16:48.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-06 16:23:31.000000000","tz":60},"subject":"Add missing string formatter","message":"Add missing string formatter\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I4e0185af5173ecec39b010eed063a95f8ef84380\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ead4a36410c4e9558707a37f84128f4783568b38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ead4a36410c4e9558707a37f84128f4783568b38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
