)]}'
{"id":"openstack%2Fneutron-lib~979433","triplet_id":"openstack%2Fneutron-lib~master~I3922689bcc0a7d0b57919fce5d9ed8d944811cd9","project":"openstack/neutron-lib","branch":"master","topic":"is_admin","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-03-09 07:25:40.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I3922689bcc0a7d0b57919fce5d9ed8d944811cd9","subject":"Use native interface to implement own is_admin check","status":"ABANDONED","created":"2026-03-08 10:51:11.000000000","updated":"2026-04-07 08:23:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4fd306194b43cf89ac43b0714279e333a07e5944","_number":979433,"virtual_id_number":979433,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-03-08 11:53:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2026-03-09 07:25:40.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-08 11:53:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-07 08:23:42.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"f2cd6d34633efdd3985b10ef01f2cdf16fa46867","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-08 10:51:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d40e210cdda64eb5a975ca66030e241af350ad6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-08 10:57:13.000000000","message":"Topic set to is_admin","accounts_in_message":[],"_revision_number":1},{"id":"3f16f2cf07cac7552dcafdcbfd7d956d7978e4f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-08 11:53:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a679594e7b54a8a94d346999aae7c60\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/270c978fa25544a3bc54a66d33845188 : SUCCESS in 1m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/279f9f752e204199b2bc3e17e742925c : SUCCESS in 8m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de5d236c80b748b6af389d891c42db10 : FAILURE in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3816cb12bd594bf4b64b47c41e3ea60c : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a824234c51a4054be8c9b9e0121a982 : FAILURE in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cac1d414573c484a86662c10d6b409c3 : FAILURE in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/142c28df10624bf9bd5d4cff19d56430 : FAILURE in 6m 09s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dda8f8d496d24734afc98b7d58080b33 : FAILURE in 15m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/789b73e612514fe1ba7ce4456d944d33 : FAILURE in 13m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c8fe2242b67840868f423de0a81c9c25 : FAILURE in 15m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7e1dc891f9994fbd935c209ab240eeb8 : FAILURE in 15m 52s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/32faad33efb9467db8b42ee9d614db1c : TIMED_OUT in 1h 01m 36s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/b1691a46265a4370a25bb0314a996787 : FAILURE in 5m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/76d5a8df98d947f8be3f45517a167a70 : FAILURE in 41m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a9787c8808ed68f0b16b627597dd58f0047ade8e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-09 07:25:40.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4fd306194b43cf89ac43b0714279e333a07e5944","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-04-07 08:23:42.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"762965d88067c5f01f4e465450b4e15080e09f6e","revisions":{"762965d88067c5f01f4e465450b4e15080e09f6e":{"kind":"REWORK","_number":1,"created":"2026-03-08 10:51:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/979433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/33/979433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/33/979433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/33/979433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/33/979433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/33/979433/1"}}},"commit":{"parents":[{"commit":"5f55e822da885f7965afecd5c83e3a9584eed367","subject":"Merge \"Bump pre-commit-config versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5f55e822da885f7965afecd5c83e3a9584eed367"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 10:50:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 10:51:00.000000000","tz":540},"subject":"Use native interface to implement own is_admin check","message":"Use native interface to implement own is_admin check\n\nThe oslo.context 6.3.0 release introduced the new native method for\nthis.\n\nDepends-on: https://review.opendev.org/974617\nChange-Id: I3922689bcc0a7d0b57919fce5d9ed8d944811cd9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/762965d88067c5f01f4e465450b4e15080e09f6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/762965d88067c5f01f4e465450b4e15080e09f6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
