)]}'
{"id":"openstack%2Fneutron-lib~982394","triplet_id":"openstack%2Fneutron-lib~master~I8300f74c4093a78498fa977d7333ad89791e546b","project":"openstack/neutron-lib","branch":"master","topic":"bug/2146542","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-03-27 18:52:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8300f74c4093a78498fa977d7333ad89791e546b","subject":"Revert \"Adds Remote Firewall Groups to FWaaS V2 Rules\"","status":"MERGED","created":"2026-03-27 11:14:47.000000000","updated":"2026-03-27 18:53:48.000000000","submitted":"2026-03-27 18:52:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982394-bug/2146542","meta_rev_id":"4293afd2583729a9df7441a3fdd88cc77685b6da","_number":982394,"virtual_id_number":982394,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-27 18:52:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2026-03-27 16:32:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2026-03-27 13:11:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2026-03-27 16:32:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-27 13:05:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-27 13:11:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-03-27 16:32:23.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"5d35ba1c82a00d76d8886c5bde9745ecb7c08e12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-27 11:14:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2bdac29e404f1e2162e459d51ec4a0057e0f9147","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 13:05:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87d7f515f9fc41e4b8e47251f60f2a23\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/315ba7ff9396416ab93bdeda69c5273f : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21c766d463d14da9810e4fac2976a3e4 : SUCCESS in 8m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca9ab0b36d924fb1aec5bd7a91603c15 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30b15d1360644fcb82abe23153748626 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2aff32ea3bcd4d21a991808f86db3ced : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2946bd070e054a30895b5973800e0163 : SUCCESS in 3m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/268804242ce048f2b59554047eede45a : SUCCESS in 6m 06s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/006a07b75e1f46cb876511bc346e3781 : SUCCESS in 1h 46m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dba1408a3918431d995cfb0bb60942fd : SUCCESS in 2m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f0859e60c30440eabe65408f39c84fd2 : SUCCESS in 1h 19m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dd95156762234e02b8a94c6a09f06fee : SUCCESS in 1h 46m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/816d0507b0834ee8ba2de777238bcdc8 : SUCCESS in 34m 26s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/34ee483d924d4f5793697094eeb58cf6 : SUCCESS in 30m 25s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/54cb632fff804423abf6b167152b7388 : SUCCESS in 3m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7ac6c1224a404aef82fe59f0d8cd9aaa : SUCCESS in 1h 28m 58s","accounts_in_message":[],"_revision_number":1},{"id":"42d4e4bfea3fa889427a0914376b9b629e095fdf","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-27 13:11:35.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d969d09b24e728c4a9f36f7925de6c8fdc5a29bb","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-03-27 16:32:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d04cf5814cb85f55afefb445361338d86670d36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 16:32:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9f5ac0d5de6e3447f3e1ddc2e2a90beb6fb6eff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 18:52:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/836f2d2e01dd4940b6f1549d05ab7682\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d237a6c7d5614c2db12d39f82f5cd965 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81c74f4ab2f94d81b70453cb2baf1af8 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869dcfdf1f1441fea548330352e70616 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a88024dd6a44ca28772cc610053bbf8 : SUCCESS in 2m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36bf045e9c4a42928e688a56f711c5f3 : SUCCESS in 6m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3de13316007545dc952cdd35bd79b5a2 : SUCCESS in 1h 50m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/519901ee515b46f1a0917eab5d19299c : SUCCESS in 3m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e72fd00468d34c8fbce553d2cc632063 : SUCCESS in 1h 37m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1fb2f6fb8ac34523bfb38f783d05abf7 : SUCCESS in 2h 19m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4f9b809506d04c818888073fbf8e5e35 : SUCCESS in 1h 22m 00s","accounts_in_message":[],"_revision_number":1},{"id":"10bd446739dd306ae95414f9fbdb8d117d03955b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 18:52:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4293afd2583729a9df7441a3fdd88cc77685b6da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 18:53:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e00dfe1bfe5146a79fdc4c53918827ee\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/bf8012fa87f64df59c30567b9d387a59 : SUCCESS in 48s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/270df2c6128641999d9a36f0dd892c55 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c12b6a314f204444b32b767d571debfb : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a97b1c15876d949553a10ba186826777a7a6526","revisions":{"6a97b1c15876d949553a10ba186826777a7a6526":{"kind":"REWORK","_number":1,"created":"2026-03-27 11:14:47.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/94/982394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/94/982394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/94/982394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/94/982394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/94/982394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/94/982394/1"}}},"commit":{"parents":[{"commit":"b1ee750c932da1b0027b0b653ea6a566742ee2b9","subject":"Add the is_filter attribute to status on trunk port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b1ee750c932da1b0027b0b653ea6a566742ee2b9"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-03-27 11:06:30.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-03-27 11:14:38.000000000","tz":60},"subject":"Revert \"Adds Remote Firewall Groups to FWaaS V2 Rules\"","message":"Revert \"Adds Remote Firewall Groups to FWaaS V2 Rules\"\n\nThis reverts commit c30700866ea32dd836493d0d0f39e6a6dad7fe0b.\n\nImplementation of those two parameters was proposed in [1] but never\nreally finished. After about 7 years of having it only in API ref I\nthink we can clearly remove this from api-ref too.\n\n[1] https://review.opendev.org/c/openstack/neutron-fwaas/+/521207\n\nCloses-bug: #2146542\nChange-Id: I8300f74c4093a78498fa977d7333ad89791e546b\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6a97b1c15876d949553a10ba186826777a7a6526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6a97b1c15876d949553a10ba186826777a7a6526"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
