)]}'
{"id":"openstack%2Fneutron-lib~982647","triplet_id":"openstack%2Fneutron-lib~master~I54505fe877f6223260ccdd493ec336618bf89f25","project":"openstack/neutron-lib","branch":"master","topic":"bug/2146531","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-04-01 20:25:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I54505fe877f6223260ccdd493ec336618bf89f25","subject":"vpnaas: add extension to remove sha1 and 3des","status":"MERGED","created":"2026-03-30 13:29:50.000000000","updated":"2026-04-01 20:26:08.000000000","submitted":"2026-04-01 20:25:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982647-bug/2146531","meta_rev_id":"d1e4bdbcbfeac1c8191a567d3848e90847b1a0b6","_number":982647,"virtual_id_number":982647,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-01 20:25:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-04-01 08:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-04-01 13:50:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-04-01 13:50:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-03-31 13:38:09.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-30 15:28:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-30 16:01:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-03-31 08:17:00.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-03-31 08:17:00.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-04-01 08:52:53.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-04-01 13:50:11.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"a2b63b744fb4fd19923adabb3345c88b3c790adc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-30 13:29:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82c4595b721e9497962ce863411560cd36b0d8db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 15:28:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2920a588e284bb3b98fa7ae072448aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/016386a7720a495289cdc6868e8b5433 : SUCCESS in 11m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d5b581cb9d84f27ae8630028294ef5d : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7549679f52a34f83bab889ef2eaf3edc : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52f59a8a4c864255bc1a582ee732acba : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d757b4f21f214bc1aa75c3216e8feead : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6320276c284243e5885a2e95ea407b3c : SUCCESS in 4m 31s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e131c1abe09741bf98b500fdbc802709 : SUCCESS in 1h 00m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65e2f169f3b54b9eb180167c1d0b1a75 : SUCCESS in 4m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/17ed02eb85b646069b5e7f7f4683724b : SUCCESS in 1h 25m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6783b98eb6804e2d92902824889289ef : SUCCESS in 1h 55m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/456f6bcb2f594111a405616bd1dd3d22 : SUCCESS in 33m 59s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/62e86d82648349529b8b1b8d43e15a05 : SUCCESS in 19m 48s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e63b326d0a6d414c98024a5381b43c5f : SUCCESS in 6m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/32a023d643a94e2d8c99c5b7bd6e20e8 : SUCCESS in 1h 22m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ada85c4eed06eddc4798fb3637ca7a97a0285d1e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-31 13:38:09.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"e81a0a4bb51019882106272b48417a0e905aa398","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-01 08:52:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"521bb751e9bae252039dc0bcb44f72aecd1a1a18","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-01 13:50:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5cbc5b3bda6267f237dd08284a6ea029b1a15d9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 13:50:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26fdfaacb289a35706b7b24ea1bfc791036bf919","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 15:49:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/643d7b15b2084c39b069fee18b056623\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5de1e16afd0344e69b56a36fc274513e : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eebcdc59cad4e52b4eff7e985671cce : FAILURE in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0cff679007fb45bbb00ee192a63051d9 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/004b97c92337400aa378bcbfebea801f : SUCCESS in 4m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a01734476e884b6eb58e2a4609a924b1 : SUCCESS in 1h 59m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c11acb45d6b4bc89f9a84567d5ddf14 : SUCCESS in 3m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9c3ca3818e3441e18bc31dcfe556596c : FAILURE in 34m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/39b67c720fb54a9cb9149b3676ac5c77 : SUCCESS in 1h 55m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dbbb5801f75d4b8685c1382daafc1afe : SUCCESS in 42m 43s","accounts_in_message":[],"_revision_number":1},{"id":"ee1255d77f4a3225561cc0f8bda8aecd24476ec3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-01 16:20:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4eb91c6a08d39e12e4d7a6116b96144fe8bcbede","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 18:33:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62e41e7234a24f41b979f108f64a15ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61eb2e536adf4e50aa65671cd8d2ca2d : SUCCESS in 9m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5dc73c684b4a434c900353db7e102276 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f997227dd4c4ce0a796db5b76e1214a : SUCCESS in 6m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f731066a649e49faaf27788de39b818c : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5de344df239a41feb4c7a7e1dff425b0 : SUCCESS in 4m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/82d8e1887c03489aa384ddce0ded043c : SUCCESS in 6m 28s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c63f63809fa04bf090fb1dd8099e3be3 : SUCCESS in 2h 07m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61ce9c2ffdc24f36b64fc24bb826b9c5 : SUCCESS in 3m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1b0f4b030a954fbe940755dd34ab0f07 : SUCCESS in 1h 32m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a9f4e75315884ea4b745cbbbbca90cde : SUCCESS in 2h 02m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f0ff639aa97543c9af4510b2d8badfd2 : SUCCESS in 38m 44s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/99aaae2eb7a74f5695b8d529f46c328c : SUCCESS in 22m 08s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/b67d1d4d63db4df6ad7973c5d6366499 : SUCCESS in 5m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/96aa69c4911745f4aede6cdcea7b57e3 : SUCCESS in 1h 24m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a862a42ab6dc4af0acc633669e6880c831ef0ec1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 18:34:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4fa42403c50b3e1eaeaaec4a235d03eb2618246","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 20:25:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f58932ca07e8479e99c1ef18c4d2169b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e98421514c9744d597f923d0654abfbc : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1b1ae5578504046bb266d1bb4a201d0 : SUCCESS in 6m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f16510da4f64fe39da127b882c2174f : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/524d9f996c0142668e29d1abeefe83f7 : SUCCESS in 4m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c66d60129da40618dadc86a03bd5031 : SUCCESS in 55m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bded747210449af8a5d1819a9421124 : SUCCESS in 4m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9312c59fe72844ec888155a0e684e8e7 : SUCCESS in 1h 49m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ba5289ce9c134a1cb4b7ad76eedc81cd : SUCCESS in 1h 49m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a78d4bbf27fa4b49a0fcb5f2e8605b95 : SUCCESS in 1h 07m 54s","accounts_in_message":[],"_revision_number":1},{"id":"581185cd0a6402e1a02bdc25069fef047a849fe1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 20:25:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d1e4bdbcbfeac1c8191a567d3848e90847b1a0b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 20:26:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af613e2e10f94c959c0890d583231ffc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d986a9733d56434493fb1842f05f9c0d : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81b9d22b34154d67a003337b704afb1e : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd6d3895b8140ce10ae4a3eab03d1153aabfab55","revisions":{"bd6d3895b8140ce10ae4a3eab03d1153aabfab55":{"kind":"REWORK","_number":1,"created":"2026-03-30 13:29:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/982647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/47/982647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/47/982647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/47/982647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/47/982647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/47/982647/1"}}},"commit":{"parents":[{"commit":"b9b52ce9b8fe1a5a02c3141355bef3055c663f46","subject":"Merge \"Revert \"Adds Remote Firewall Groups to FWaaS V2 Rules\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b9b52ce9b8fe1a5a02c3141355bef3055c663f46"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-30 13:19:20.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-30 13:28:53.000000000","tz":120},"subject":"vpnaas: add extension to remove sha1 and 3des","message":"vpnaas: add extension to remove sha1 and 3des\n\nAdd a new vpn-no-sha1-3des API extension that removes the\ndeprecated sha1 authentication algorithm and 3des encryption\nalgorithm from VPN IKE and IPsec policy choices.\n\nWhen this extension is loaded, the default auth_algorithm\nchanges from sha1 to sha256 and the supported algorithm\nlists exclude the insecure legacy algorithms.\n\nThe extension follows the same pattern as vpn-aes-ctr,\noverriding the RESOURCE_ATTRIBUTE_MAP for both ikepolicies\nand ipsecpolicies resources.\n\nRelated-Bug: #2146531\n\nAssisted-By: claude-4.6-opus-high\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I54505fe877f6223260ccdd493ec336618bf89f25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bd6d3895b8140ce10ae4a3eab03d1153aabfab55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bd6d3895b8140ce10ae4a3eab03d1153aabfab55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
