)]}'
{"id":"openstack%2Fneutron-lib~983203","triplet_id":"openstack%2Fneutron-lib~master~I3228784a12402c76c58a333e897ec2fd56e2b266","project":"openstack/neutron-lib","branch":"master","topic":"aap-reject-multicast","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2026-04-14 17:26:42.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-04-09 17:07:40.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I3228784a12402c76c58a333e897ec2fd56e2b266","subject":"Reject multicast addresses in AAP: n-lib part","status":"MERGED","created":"2026-04-02 13:25:18.000000000","updated":"2026-04-14 17:28:05.000000000","submitted":"2026-04-14 17:26:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"983203-aap-reject-multicast","meta_rev_id":"c7ab74ca921301775cb3471d6a9e2ae46e4f7c43","_number":983203,"virtual_id_number":983203,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-14 17:26:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-04-13 06:08:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-04-09 17:07:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-04-14 15:25:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-04-13 06:08:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-02 15:10:16.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-04-02 16:21:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-02 16:36:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-04-09 17:07:40.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-04-13 06:08:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"6948ca8e5b14452e0769cc66a857d8450fa5b904","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-04-02 13:25:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c4323fb7cffecae93f835b99339e3301bd889d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 16:21:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cec2b995998b44d8b8b5121012dc9d6d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eda556aef33c47a1976c6ffc90c4af8f : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b973e5b6f3a48379c8d6ee6cabb47a7 : SUCCESS in 4m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b6895781bdf419f8a6b769a00cc05ca : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51994b335ce5411087d2e33a8bd45014 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4388d5047e04eacb24d01fa5b0560fa : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d562d03b95504272bfe65268e601aaa5 : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4947a0d21262451297fd7f75afdf12a5 : SUCCESS in 8m 31s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f350e0d2a9d74dfab695ed7cdc2e9f40 : SUCCESS in 1h 51m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cd5f4fce32d2468aad0f3353484f039a : TIMED_OUT in 2h 49m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dad5ebd9cbce4622b8fb6adbc53476b8 : SUCCESS in 2h 02m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2949ad1457d74a70b43d81187da7832a : SUCCESS in 51m 08s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/c4d20b871c754915b474c2947dcc8e83 : SUCCESS in 19m 24s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a805f6b84faa4590abc53d485660d4ba : SUCCESS in 5m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/38cd4ec1ecb74bf0a991c214e28840bb : SUCCESS in 1h 23m 01s","accounts_in_message":[],"_revision_number":1},{"id":"040239f0fd49377b8d0261b068935f82377eea75","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-02 16:36:33.000000000","message":"Patch Set 1:\n\n(9 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e35457b38daaee68e3aa3ef5a1ebf8e95a7e3c7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-04-07 13:22:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"64d6f8ab981a3f818d52c3dbbf5526c6e20ec7f6","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-04-07 13:23:15.000000000","message":"Patch Set 1:\n\n(10 comments)","accounts_in_message":[],"_revision_number":1},{"id":"def429a5db6a45a458490ea73e396d4ff96900d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 15:35:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c2865657cf8438ebd65610f63375226\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a4be888f5a6e4e69a887a4f84ba6e038 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7737cefcb06d40799c0b54afb9c80419 : SUCCESS in 8m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c5c4dde543a4ea2a45adf83eeec253a : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ae28f69fc9d49328af94a49c22adaa3 : FAILURE in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d859b1eeeb047fa9dc54937bf1e1e5c : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fae1f5b1ae0d493ca1e6a77cfdfcc1d8 : SUCCESS in 4m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ecdba8dbea47470cb049733b6025e08a : SUCCESS in 8m 23s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f619881def8454d9fadb3f26a396767 : SUCCESS in 1h 47m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e2fdcf04191f47d3855271473f0b729b : SUCCESS in 1h 16m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f8f95df5599b4310ab6316e5d10ecafc : SUCCESS in 1h 58m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4793716da831475bab149eed1d08cee5 : SUCCESS in 33m 32s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/e86c083e6eda4cd9b305067934a7c86e : SUCCESS in 21m 40s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d8ccf37c0eae4050ae8885ec8fc46f96 : SUCCESS in 5m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/87320547d3ed48ed81ad3796d361edc4 : SUCCESS in 1h 20m 40s","accounts_in_message":[],"_revision_number":2},{"id":"2ccf1bb7200331d72c8a580fbc4f2669d6379f9a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-07 19:59:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b0cf2585213a5a8b9c39ab2f5387af94f785dd06","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-04-08 09:44:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c27558eaecfaebc2b1fefd977cceb2ed6290f3b2","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-04-08 09:45:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8eeb81a0e7e068760cf91df454d048400c9c5fac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 11:42:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d44f52c312347a898d5e5b15fe484ce\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6ba854572adc4ba49be1b303b4c3007d : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce1a1536568424db7915dff11c74d94 : SUCCESS in 6m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c5ce2044a994c77ab8dbbc00c50b88a : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe03fe88c143400b87019ae2166954d4 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6001d03bc5874424afda37d72928b115 : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/820d4c456d2d4a5f81b3d2ffee439e7b : SUCCESS in 3m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/405558f9db4e40a7872340591851ebe9 : SUCCESS in 4m 59s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd06e5879eac4b5aa1d8bb4d65b61d43 : SUCCESS in 1h 41m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5c7cac7823fe4efaab8dd582931b671d : SUCCESS in 1h 37m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8ea4b3f57f4e4836b49ae10583c711d4 : SUCCESS in 1h 56m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a91c55ad30fa4942bb24f2f194ac164c : SUCCESS in 34m 06s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/8800307f74e3420391c56ca62c548260 : SUCCESS in 26m 46s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/838553ad2fc341b493ebd7d4629b3d87 : SUCCESS in 3m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/97989c7923e74af78ce509a82cefc86f : SUCCESS in 1h 23m 19s","accounts_in_message":[],"_revision_number":2},{"id":"8c5a27b0abef604450c2db7386a8577f8880ae9d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-09 17:07:40.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"17969402adf47a592fb3f54d89daaec55bdef9c0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-13 06:08:47.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"0cb0e4698f78f77bcfd2cfbbb8c94b36a5c1ab21","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-14 15:25:15.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"84288a07f02fb0cf6db02f749147b6e1f1b4d70d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 15:25:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b6eec48b5df440a8b8c01b851cd9f4585d9851f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 17:26:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a80dc7f1d7244939e4b83c98ecea760\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cb1d1726195b413b96b4117aea3701f9 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f257cada5e954c87994f271e0591371d : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b02a27e6fcac4a0d89d8f71f961cb528 : SUCCESS in 11m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0a7ea45975544f182b868e0d3858805 : SUCCESS in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a9cba1002da422e9bec4fe651ec5bca : SUCCESS in 2m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e6b3d5ae49243a9ba94ffba20d026e1 : SUCCESS in 2h 00m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/396e6f3b4ab84b8bb6d5d528337ce187 : SUCCESS in 1h 38m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b39e89bb5dd04d86896ff261434fa924 : SUCCESS in 1h 42m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/09a88fcabf1b413e964a5ed0a93843ed : SUCCESS in 1h 17m 57s","accounts_in_message":[],"_revision_number":2},{"id":"66a64e2efa2f0cf42802ca311a8ed1f88bfc0a8b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 17:26:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c7ab74ca921301775cb3471d6a9e2ae46e4f7c43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 17:28:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d57b671c135346af903c961f25fb3d53\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/8cb2f300139442838b91289004146816 : SUCCESS in 58s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740285b61e0d49f28d17f6011275b03f : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"16c421ba99f77a5fd7ab6712b7ac7a734004dc92","revisions":{"252fdeeddbd4620b5c3e7c8362f3d9b84e76528f":{"kind":"REWORK","_number":1,"created":"2026-04-02 13:25:18.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/03/983203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/03/983203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/983203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/983203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/983203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/03/983203/1"}}},"commit":{"parents":[{"commit":"b9b52ce9b8fe1a5a02c3141355bef3055c663f46","subject":"Merge \"Revert \"Adds Remote Firewall Groups to FWaaS V2 Rules\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b9b52ce9b8fe1a5a02c3141355bef3055c663f46"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-03-30 12:31:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-04-02 13:24:39.000000000","tz":120},"subject":"Reject multicast addresses in AAP: n-lib part","message":"Reject multicast addresses in AAP: n-lib part\n\nPlease see the commit message of the neutron part:\nhttps://review.opendev.org/c/openstack/neutron/+/983201\n\nChange-Id: I3228784a12402c76c58a333e897ec2fd56e2b266\nNeeded-By: https://review.opendev.org/c/openstack/neutron/+/983201\nPartial-Bug: #2147115\nSigned-off-by: Bence Romsics \u003cbence.romsics@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/252fdeeddbd4620b5c3e7c8362f3d9b84e76528f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/252fdeeddbd4620b5c3e7c8362f3d9b84e76528f"}]},"branch":"refs/heads/master"},"16c421ba99f77a5fd7ab6712b7ac7a734004dc92":{"kind":"REWORK","_number":2,"created":"2026-04-07 13:22:49.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/03/983203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/03/983203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/983203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/983203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/983203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/03/983203/2"}}},"commit":{"parents":[{"commit":"6cd11d272fbe8f427e9e2f7032d0b51b69b32330","subject":"Merge \"vpnaas: add extension to remove sha1 and 3des\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6cd11d272fbe8f427e9e2f7032d0b51b69b32330"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-03-30 12:31:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-04-07 13:22:18.000000000","tz":120},"subject":"Reject multicast addresses in AAP: n-lib part","message":"Reject multicast addresses in AAP: n-lib part\n\nPlease see the commit message of the neutron part:\nhttps://review.opendev.org/c/openstack/neutron/+/983201\n\nChange-Id: I3228784a12402c76c58a333e897ec2fd56e2b266\nNeeded-By: https://review.opendev.org/c/openstack/neutron/+/983201\nPartial-Bug: #2147115\nSigned-off-by: Bence Romsics \u003cbence.romsics@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/16c421ba99f77a5fd7ab6712b7ac7a734004dc92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/16c421ba99f77a5fd7ab6712b7ac7a734004dc92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
