)]}'
{"id":"openstack%2Fneutron-lib~983214","triplet_id":"openstack%2Fneutron-lib~master~I36be6d02a531ec4ba9876d9b35871965e8e3d456","project":"openstack/neutron-lib","branch":"master","topic":"bug/2143895","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-04-15 11:32:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I36be6d02a531ec4ba9876d9b35871965e8e3d456","subject":"Rehome neutron policy check strings to neutron-lib","status":"MERGED","created":"2026-04-02 14:45:56.000000000","updated":"2026-04-15 11:34:04.000000000","submitted":"2026-04-15 11:32:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983214-bug/2143895","meta_rev_id":"e3e15f90d06c5fd9ef5e430de6d55677fd3fe384","_number":983214,"virtual_id_number":983214,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-15 11:32:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-04-15 07:49:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-04-15 08:20:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-04-15 09:02:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-04-15 07:49:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-02 14:57:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-04-02 16:33:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-06 09:49:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-04-10 09:37:33.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"CC"},{"updated":"2026-04-15 07:49:25.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-04-15 08:20:53.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"7f55809d164c02b5aeacb388ef6c3bc3c5cc6048","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-04-02 14:45:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e69162c537e2b1c594dfaa274634de2dbc83828","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 16:33:18.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/445ed867cda04444a48344f22e411d33\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17de7c2bd79740949214924c37d5d81c : FAILURE in 7m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28fddb385753482ab1d605d2f196b797 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8abaf3b90cf4f88884298c135448d9b : FAILURE in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5ce0352c5084b5d9ac02c286772c825 : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd83108638394d418275a43640914642 : SUCCESS in 6m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9e4d0ececbe545a3a73dd66cd4e553fe : SUCCESS in 4m 50s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/607e9e5e9c6d424a8b569592436ffb53 : SUCCESS in 1h 45m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c12f515c29f34131847b8a05542807b6 : SUCCESS in 2m 10s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2b148a22b1af46fb83298b611a57314e : SUCCESS in 1h 24m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0096c083895443fd9f039d0997125e23 : SUCCESS in 1h 43m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/74658d7c9f064ab2ab138fbc93f0b37e : SUCCESS in 33m 19s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/ea2dae37a96346bca44769b7cf331928 : SUCCESS in 26m 24s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/33114f27bb9f48ec9a0e793914768001 : SUCCESS in 8m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bebbe683b4b74d35814e05c50f5f83f3 : SUCCESS in 1h 24m 45s","accounts_in_message":[],"_revision_number":1},{"id":"db7b407c75b7681a5c2ab9a95e8c04885a3d31c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-04-14 10:25:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"475a2a5428392f4b4bcccc0408db8f6751b926c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 12:50:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46006814159c4311bd8eaa9d22807c2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a22f398693c2456ab8e2cf3694f57091 : SUCCESS in 8m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/531df08b61c240129be3973648c5ded3 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f867bfe651784f37b7d4d9a6c7332e79 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c34c8604b90241018a045558a16dfe7e : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7584afa8c5149469d60711e49444863 : SUCCESS in 4m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d8bc96f1ad5c4f1bb1e6fbb36e14b959 : SUCCESS in 7m 04s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8407dd47f114957b1539377d3aa98b3 : SUCCESS in 1h 35m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/411dbd9cdf1f4b35ac5505ec0727a27e : SUCCESS in 3m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b89a003274814e47a55b04bad93bc877 : SUCCESS in 1h 34m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e6b1d2fd0db14f35a98de64da22d1db1 : SUCCESS in 2h 15m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c3bec87dca0b4912b47833027a9c55a8 : SUCCESS in 34m 46s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/6ca438397f5442a28f82b8433413a7ae : SUCCESS in 26m 10s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3a128081c0554151901257a5f7d5549f : SUCCESS in 6m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bf3553003add4fd783e20546bfccdd0e : FAILURE in 1h 29m 18s","accounts_in_message":[],"_revision_number":2},{"id":"c16bd1807341f421c5572cf6c826e6221f59099e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-04-15 07:21:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"efec93ac8d668effdd716c1c0b6f55a21961a868","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-15 07:49:25.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9dc690a86529becc5b7d22c1e686a8cdeff10c84","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-15 08:20:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fb9c453a2aedaf2bf08d5c23f0bc660d4c4a3351","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-15 09:02:28.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"056750dc9954bb1abe02676fed61383a83b86889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 09:31:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3017d24bbd6545f8b00ad77eaed64b5c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80d11129620f4daeafef682e090f60b6 : SUCCESS in 9m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75ac1ecf1743476683d8efa322a1c9eb : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aa52be0df714dd58bcd0c573d5349be : SUCCESS in 7m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f119dcc647a746edb17ba6edde93761f : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b814fd13b5f14ad68d635e50705ab417 : SUCCESS in 2m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/02c5e28ad59744ee8d29d6c7a8d38e41 : SUCCESS in 6m 02s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6c6218d48c84359b70ddb70d9192656 : SUCCESS in 1h 51m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d36ecf4489c4ad9b18e79be2c752284 : SUCCESS in 4m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c838d3bda06b4a07b9e60b9bc9aeeda4 : SUCCESS in 1h 49m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/80202d4d6f5c45c4a0488f9a43f3d2d2 : SUCCESS in 2h 09m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a6d9bdd4ac5f4ac987762b1eabe54497 : SUCCESS in 31m 43s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/0e8e49b6c502439aad5445ca35caa402 : SUCCESS in 25m 58s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/dd0b1a8daa2644d8bdbacd97eab59bd3 : SUCCESS in 3m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3e0f75bd01454ed3a659d1a6a6910c54 : SUCCESS in 1h 25m 06s","accounts_in_message":[],"_revision_number":2},{"id":"fdc5b3c754dc96739016f56b6a2ca2f575178db0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 09:31:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"03fd9b923be1decbafb62b58cf691b99741dc9e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 11:32:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5d63add1a344222b78321e0cd7c078a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96a52c22b455409a9a3a6b78ed1306e0 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc141b15d1824e0491ae259f6162c245 : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/816140cc9fe84cd8961dfd3de1ddb1cd : SUCCESS in 2m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f479e96495f4794b5664bf1b261acf9 : SUCCESS in 6m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec426cb8180c44ebb69a411dd851ba08 : SUCCESS in 1h 53m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5e46ad1ea864a6db603433a0a04d07f : SUCCESS in 3m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f58154aab20b43049f3cd461e0f4eaec : SUCCESS in 1h 34m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ab36596935454195a5f9b66f03736913 : SUCCESS in 2h 00m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/946aa87e51044eae93fc110e0c9799b8 : SUCCESS in 1h 25m 31s","accounts_in_message":[],"_revision_number":2},{"id":"eeddc878a62f78b1bede07289c12d3aeb04c4818","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 11:32:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e3e15f90d06c5fd9ef5e430de6d55677fd3fe384","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 11:34:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15b03ab5533441d7833e69b0e9d28766\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2178f7fbd374150a2f75a4a1bb957ec : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed5f9ca3d8cb4e32952b7e2bb43de36b : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bc911e231e7eb4aaf411f7d463d59c29c327c310","revisions":{"e1357d4d8f29172d37581cbca38eff6bcb48395b":{"kind":"REWORK","_number":1,"created":"2026-04-02 14:45:56.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/14/983214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/14/983214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/14/983214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/14/983214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/14/983214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/14/983214/1"}}},"commit":{"parents":[{"commit":"6cd11d272fbe8f427e9e2f7032d0b51b69b32330","subject":"Merge \"vpnaas: add extension to remove sha1 and 3des\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6cd11d272fbe8f427e9e2f7032d0b51b69b32330"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-04-02 14:08:04.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-04-02 14:19:38.000000000","tz":120},"subject":"Rehome neutron policy check strings to neutron-lib","message":"Rehome neutron policy check strings to neutron-lib\n\nThis patch moves to neutron-lib some policy check strings which were\ndefined in the neutron.conf.policy.base module but are used also by\nstadium projects so it is better to have them in neutron-lib.\n\nRelated-bug: #2143895\n\nChange-Id: I36be6d02a531ec4ba9876d9b35871965e8e3d456\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e1357d4d8f29172d37581cbca38eff6bcb48395b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e1357d4d8f29172d37581cbca38eff6bcb48395b"}]},"branch":"refs/heads/master"},"bc911e231e7eb4aaf411f7d463d59c29c327c310":{"kind":"REWORK","_number":2,"created":"2026-04-14 10:25:16.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/14/983214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/14/983214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/14/983214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/14/983214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/14/983214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/14/983214/2"}}},"commit":{"parents":[{"commit":"6cd11d272fbe8f427e9e2f7032d0b51b69b32330","subject":"Merge \"vpnaas: add extension to remove sha1 and 3des\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6cd11d272fbe8f427e9e2f7032d0b51b69b32330"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-04-02 14:08:04.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-04-14 10:25:04.000000000","tz":120},"subject":"Rehome neutron policy check strings to neutron-lib","message":"Rehome neutron policy check strings to neutron-lib\n\nThis patch moves to neutron-lib some policy check strings which were\ndefined in the neutron.conf.policy.base module but are used also by\nstadium projects so it is better to have them in neutron-lib.\n\nRelated-bug: #2143895\n\nChange-Id: I36be6d02a531ec4ba9876d9b35871965e8e3d456\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bc911e231e7eb4aaf411f7d463d59c29c327c310"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bc911e231e7eb4aaf411f7d463d59c29c327c310"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
