)]}'
{"id":"openstack%2Fneutron-lib~984135","triplet_id":"openstack%2Fneutron-lib~master~I75f327792a41baa34c75ece8281c0e6baf390b05","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-05-11 14:13:38.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-05-13 08:35:45.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-05-13 10:46:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I75f327792a41baa34c75ece8281c0e6baf390b05","subject":"Create name string validator","status":"MERGED","created":"2026-04-10 22:11:02.000000000","updated":"2026-05-13 10:47:28.000000000","submitted":"2026-05-13 10:46:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984135","meta_rev_id":"e8761d0c03fafef137b02d864560b8e9de07c7fe","_number":984135,"virtual_id_number":984135,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-13 10:46:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-05-11 14:13:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-05-13 08:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-05-13 08:35:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-05-11 12:26:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-10 23:46:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-14 06:08:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-04-29 07:01:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-04-29 07:01:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-04-29 07:01:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-05-04 10:22:22.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"bb9c3e09242343a8acfeb76494d3617fa9c1f55f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-10 22:11:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8add3a85e207366fdbb8f00ed051fd84dc2e5017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 23:46:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36733354f0f948d89b426d8994976a8c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daf328405d4345f681d3d65c2c033368 : SUCCESS in 9m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/747b8bd8a11b4e10a8207460fe926caf : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19598d8df49647e38ee3248b97b39852 : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a451d7278de248ca81bda1ed3aa18745 : SUCCESS in 2m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/145c96bf671542f9a08b4198fb5d4483 : SUCCESS in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3547740a92b54094ac102ac99a306a67 : SUCCESS in 5m 51s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/303be67a1f274d7283feeeb39b0aeb6e : SUCCESS in 1h 27m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d98e8f15120f434eb020ccd9c3a60bd3 : SUCCESS in 1h 20m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1479698746a3460ba18058f8ecefe543 : SUCCESS in 1h 34m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5361152f3ef549e4b5cc8cbe9e6bdac7 : SUCCESS in 33m 07s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/ad748e1ed55448668ecb1dfab61d5ce9 : SUCCESS in 22m 45s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/4d27c6a58cc647cda0a3a70eda710636 : SUCCESS in 6m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fd17b2e77fec468c82d0fc62f9e6c98b : SUCCESS in 1h 21m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c8c29ffc76ed3cd34afc96831383c80b7c2195a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-14 02:33:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"1df202821efb2d12e1e5faa2ade5e819284f8f52","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-14 02:33:48.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"1144f192c601f7c33952b1e98585b1bf42973113","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-14 03:32:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa7f8e79a58965e294504730cff2ef373c45918b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-14 04:16:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1fdb15fed40ef5ad28e84fc89ebba16f3cbfb073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 06:29:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f3519d07c7147a3b385a4311e35221f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c13df8881b94442ac4eb86e64dd5777 : SUCCESS in 4m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b601383042fd4299b0e13bb457f6f4ac : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3629c6f1c4194647a5310d7b40f10042 : SUCCESS in 8m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0352d2246da948f69898fd7db81ab560 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/28f809e1d96c413eb2fbbbedda69f0fe : SUCCESS in 3m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1b9ead2ce15044759c2469a272c7ab6a : SUCCESS in 5m 51s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a20680c4bc5b47059e5b7631a6cec2d7 : SUCCESS in 1h 46m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/27c22fe1c4fb4ca0a89ca26da7646eff : SUCCESS in 1h 16m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a77a248c6b624f16a88102496c712813 : SUCCESS in 2h 12m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7d412be442bb4682a433a98cb3d4d67b : SUCCESS in 30m 32s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/0342ad458ff149e495708c7d3a9adad3 : SUCCESS in 22m 35s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9c7f879268ae4513ab9a313971c35a4b : SUCCESS in 14m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/43c920b29fc5486fa6b4f0e66b1dff6f : SUCCESS in 1h 20m 03s","accounts_in_message":[],"_revision_number":4},{"id":"eef84e098b47d0446892fbdce62654bb7b0a7f07","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-15 13:08:07.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"12e1ea0c79a399ff2e6c52989d945f60e6c40cca","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-25 01:44:35.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"4fdc16c4f93cc01d344668a59be03bd18ae87c3e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-29 07:01:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a480b5a325802fd74da4a43fb0c7af18b549f118","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-01 23:05:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7fd74ca49a246ca136a34730fddede598129736e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-01 23:05:29.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"00752cd97bff978623adccef5e77dcf7698c5b03","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-01 23:12:41.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"8ded06c7379a04bfc75d0683636ab655e3781d22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 01:39:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9604d159974743c99861834fc5dbdc6a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/374d9211834c448ead57245c8dfdb880 : SUCCESS in 7m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29b745791728410e8645aae20db99003 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4738270daf794f77bc7f71311de29e19 : SUCCESS in 6m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79e8838aecdd406e8d377b5368d9889f : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be68812de7ee4f28bb77210f80ff5f81 : SUCCESS in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6e4bdf5f040e401194be3ff484e343d7 : SUCCESS in 5m 18s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eecf3adf06824600847d52cc00545a07 : SUCCESS in 1h 48m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6975771e573c48deb5d181caa705b047 : SUCCESS in 1h 36m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/332096cbd89e4ec798b847d6189ffb4d : SUCCESS in 2h 32m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/552b9673158b40a5bc656d428ee204a3 : SUCCESS in 31m 21s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/61f1a1fd086c4382bff0fe8744329c71 : SUCCESS in 25m 19s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5759df647f8941bf8b2d46a022af7e27 : SUCCESS in 5m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/23b05db1038a476fbe62eef3d6e60d1f : SUCCESS in 42m 06s","accounts_in_message":[],"_revision_number":5},{"id":"39e50737d6f1f2edf746061665ddf2b32e73d31b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-04 10:22:22.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"595800aa9882a35c0d11c24c193a1242ef27c566","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-11 11:16:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e08bebd2785025fae3906c92ada1bd05565cb66d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-11 12:22:28.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"552f2f8bdccfaf1549db9f771e842863bb13016f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-11 12:23:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"16c3c308222f92cb6a283283b5c2e41dff1e5d91","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-11 12:26:10.000000000","message":"Patch Set 6: Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"6807f6816ed244b493165570daf1606cb312d998","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-11 14:13:38.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"91d13b14244af3ed05aa0b4d45a5d5a92e638b27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 14:14:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be251c2c88604e8bb7a737dda815928f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9910812b83f14ffda42927f87654fdf8 : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2813cb308ec460ca6b3118b57b74d34 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12cd49c3c1ac4915a3fdf7545ec7d459 : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35d7102d9ba342ce9c0fca58477843fc : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1ab6d4042f2411d994ab8297fdcc7e2 : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/20356a9f19d14721b01e86b914d54542 : SUCCESS in 3m 33s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d626e2959208452da3014f58f483d524 : SUCCESS in 1h 46m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e576417856ca4eaeac67c35717fac455 : SUCCESS in 1h 33m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cf33089dd59441efb5015e52a17226f5 : SUCCESS in 1h 39m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/98784d39ce3f4af0aad6390a1f35d55e : SUCCESS in 36m 52s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/01d6f25fafff4e8fb01694852359a995 : SUCCESS in 16m 16s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/48c76d44d71f476baad82a7778fd1862 : SUCCESS in 6m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5b27acb90d96496fa3385f39d755ccde : SUCCESS in 1h 28m 21s","accounts_in_message":[],"_revision_number":6},{"id":"c97880e66277bdb23e0fc6ff139358ad4c213c97","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-13 08:35:45.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"21ae45afbff394f25a3d007e6b52884077f97b05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 08:36:14.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"116e9b1d487d0419b3963fd3e6d23f0443ef189b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 10:46:06.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6069a3a87dbd42c38c7659fb6fd2d428\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/321e0203283346c58e1c3cf705ea3967 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb233dfd01eb4b59a7fbf11d67a9f6f9 : SUCCESS in 6m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/48f1a2826cc04357aa49e1ff78002aef : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f430b323c2934e80af9c210b225f91eb : SUCCESS in 9m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6cbab72f91344613b979625a2d747b43 : SUCCESS in 2h 02m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fc912b8a7b345c0b41d7fb9346f0fdf : SUCCESS in 5m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cd6187064711447b98723ac45326d22b : SUCCESS in 1h 28m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5fdb416d9cc74039b8b3435681147b56 : SUCCESS in 1h 51m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/253a04ee519640d6930318754a346163 : SUCCESS in 1h 12m 40s","accounts_in_message":[],"_revision_number":6},{"id":"c759f8e1cfe8fb48247999bb3aae753a6abbc46d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 10:46:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"e8761d0c03fafef137b02d864560b8e9de07c7fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 10:47:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/801a3438f5e842c8a160ea54d92c7378\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b003e4fc9ad4530adc70d6ab2453002 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6ca7b3a89ccba4bd59550a94c6a89345be2afa24","revisions":{"e79df97ecf1d35be123a412c85325ce99044ffe9":{"kind":"REWORK","_number":1,"created":"2026-04-10 22:11:02.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/35/984135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/35/984135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/1"}}},"commit":{"parents":[{"commit":"6cd11d272fbe8f427e9e2f7032d0b51b69b32330","subject":"Merge \"vpnaas: add extension to remove sha1 and 3des\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6cd11d272fbe8f427e9e2f7032d0b51b69b32330"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-10 22:05:40.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-10 22:05:40.000000000","tz":-240},"subject":"WIP: Create name string validator","message":"WIP: Create name string validator\n\nSending an emoji as a name for a neutron resource can\nresult in a 500 error. Create a more restrictive\nstring validator for names.\n\nBased on the Nova code at\nnova/api/validation/parameter_types.py\n\nTODO: use it somewhere\n\nCloses-bug: #2095097\n\nChange-Id: I75f327792a41baa34c75ece8281c0e6baf390b05\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e79df97ecf1d35be123a412c85325ce99044ffe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e79df97ecf1d35be123a412c85325ce99044ffe9"}]},"branch":"refs/heads/master"},"8bad40bebdd1dc67188eff14db9a8cc757dd2ad5":{"kind":"REWORK","_number":2,"created":"2026-04-14 02:33:35.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/35/984135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/35/984135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/2"}}},"commit":{"parents":[{"commit":"6cd11d272fbe8f427e9e2f7032d0b51b69b32330","subject":"Merge \"vpnaas: add extension to remove sha1 and 3des\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6cd11d272fbe8f427e9e2f7032d0b51b69b32330"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-10 22:05:40.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-14 02:30:59.000000000","tz":-240},"subject":"Create name string validator","message":"Create name string validator\n\nSending an emoji as a name for a neutron resource can\nresult in a 500 error. Create a more restrictive\nstring validator for names, as well as one to validate\nthe name string is not empty.\n\nChanged all the object validators to start using them.\n\nBased on the Nova code at\nnova/api/validation/parameter_types.py\n\nCloses-bug: #2095097\n\nChange-Id: I75f327792a41baa34c75ece8281c0e6baf390b05\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8bad40bebdd1dc67188eff14db9a8cc757dd2ad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8bad40bebdd1dc67188eff14db9a8cc757dd2ad5"}]},"branch":"refs/heads/master"},"3d16121b888dc879e6c109bf75371883ca547855":{"kind":"REWORK","_number":3,"created":"2026-04-14 03:32:27.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/35/984135/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/35/984135/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/3"}}},"commit":{"parents":[{"commit":"6cd11d272fbe8f427e9e2f7032d0b51b69b32330","subject":"Merge \"vpnaas: add extension to remove sha1 and 3des\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6cd11d272fbe8f427e9e2f7032d0b51b69b32330"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-10 22:05:40.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-14 03:32:10.000000000","tz":-240},"subject":"Create name string validator","message":"Create name string validator\n\nSending an emoji as a name for a neutron resource can\nresult in a 500 error. Create a more restrictive\nstring validator for names, as well as one to validate\nthe name string is not empty.\n\nChanged all the object validators to start using them.\n\nBased on the Nova code at\nnova/api/validation/parameter_types.py\n\nCloses-bug: #2095097\n\nChange-Id: I75f327792a41baa34c75ece8281c0e6baf390b05\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3d16121b888dc879e6c109bf75371883ca547855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3d16121b888dc879e6c109bf75371883ca547855"}]},"branch":"refs/heads/master"},"287034c89a0c2fd6588f01715816ad438083468c":{"kind":"REWORK","_number":4,"created":"2026-04-14 04:16:18.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/35/984135/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/35/984135/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/4"}}},"commit":{"parents":[{"commit":"6cd11d272fbe8f427e9e2f7032d0b51b69b32330","subject":"Merge \"vpnaas: add extension to remove sha1 and 3des\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6cd11d272fbe8f427e9e2f7032d0b51b69b32330"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-10 22:05:40.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-14 04:13:53.000000000","tz":-240},"subject":"Create name string validator","message":"Create name string validator\n\nSending an emoji as a name for a neutron resource can\nresult in a 500 error. Create a more restrictive\nstring validator for names, as well as two other\nvariants to validate the name string is not empty\nor None.\n\nChanged all the object validators to start using them.\n\nBased on the Nova code at\nnova/api/validation/parameter_types.py\n\nCloses-bug: #2095097\n\nChange-Id: I75f327792a41baa34c75ece8281c0e6baf390b05\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/287034c89a0c2fd6588f01715816ad438083468c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/287034c89a0c2fd6588f01715816ad438083468c"}]},"branch":"refs/heads/master"},"8ca2f498a3a977670a03de6de5702a90a36266eb":{"kind":"REWORK","_number":5,"created":"2026-05-01 23:05:29.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/35/984135/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/35/984135/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/5"}}},"commit":{"parents":[{"commit":"fe86c1988a2fed008de27b055d0fd127daffaf2a","subject":"Remove logging-resource API definition and documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fe86c1988a2fed008de27b055d0fd127daffaf2a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-10 22:05:40.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 23:04:40.000000000","tz":-240},"subject":"Create name string validator","message":"Create name string validator\n\nSending an emoji as a name for a neutron resource can\nresult in a 500 error. Create a more restrictive\nstring validator for names, as well as two other\nvariants to validate the name string is not empty\nor None.\n\nChanged all the object validators to start using them.\n\nBased on the Nova code at\nnova/api/validation/parameter_types.py\n\nCloses-bug: #2095097\n\nChange-Id: I75f327792a41baa34c75ece8281c0e6baf390b05\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8ca2f498a3a977670a03de6de5702a90a36266eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8ca2f498a3a977670a03de6de5702a90a36266eb"}]},"branch":"refs/heads/master"},"6ca7b3a89ccba4bd59550a94c6a89345be2afa24":{"kind":"REWORK","_number":6,"created":"2026-05-11 12:22:28.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/35/984135/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/35/984135/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/35/984135/6"}}},"commit":{"parents":[{"commit":"e3dacf2eb28760a28138014b7202ed2f816b2cd5","subject":"Merge \"Remove router-interface-fip API definition and documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e3dacf2eb28760a28138014b7202ed2f816b2cd5"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-10 22:05:40.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-11 12:19:03.000000000","tz":120},"subject":"Create name string validator","message":"Create name string validator\n\nSending an emoji as a name for a neutron resource can\nresult in a 500 error. Create a more restrictive\nstring validator for names, as well as two other\nvariants to validate the name string is not empty\nor None.\n\nChanged all the object validators to start using them.\n\nBased on the Nova code at\nnova/api/validation/parameter_types.py\n\nCloses-bug: #2095097\n\nChange-Id: I75f327792a41baa34c75ece8281c0e6baf390b05\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6ca7b3a89ccba4bd59550a94c6a89345be2afa24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6ca7b3a89ccba4bd59550a94c6a89345be2afa24"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
