)]}'
{"id":"openstack%2Fneutron-lib~986579","triplet_id":"openstack%2Fneutron-lib~master~I9755da5ff7912b35e78f782a6d2edb8992279682","project":"openstack/neutron-lib","branch":"master","topic":"bug/2144617","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-05-01 13:54:22.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"38298":{"account":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"last_update":"2026-05-02 01:05:43.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-05-02 01:05:43.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-05-02 01:05:43.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2026-05-02 01:05:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9755da5ff7912b35e78f782a6d2edb8992279682","subject":"evpn: Add EVPN API definitions","status":"MERGED","created":"2026-04-28 20:39:16.000000000","updated":"2026-05-02 01:06:38.000000000","submitted":"2026-05-02 01:05:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986579-bug/2144617","meta_rev_id":"0cc2f1647011b2cf48e287e35cb1a7daa9c63ab3","_number":986579,"virtual_id_number":986579,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-02 01:05:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-05-01 13:54:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-05-01 22:36:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-05-01 22:36:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-05-01 13:54:22.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-28 23:16:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-29 07:52:53.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-04-29 15:33:45.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2026-04-29 15:33:45.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"state":"REVIEWER"},{"updated":"2026-04-29 15:33:45.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2026-05-01 22:36:45.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"62f7ae4c4fb2846d12cc93a26c349f173ededae1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-28 20:39:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a20e3461867e94c5271722245f4bf429b7b85c36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 23:16:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b05a896b07524402832779ccc718dfd2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72244cba8b1f41ea98e68c9ff47c8080 : SUCCESS in 8m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93922c3ae882468d95d4c8d34881804d : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0107b1d99d6c44cd8d16fd03812353fd : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/284efce610e446629ba855eceddc20f2 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/57c92dc2d56845479e2165d486da4c46 : SUCCESS in 3m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9a0f1583dd684bc68b18d3a6e341462c : SUCCESS in 5m 52s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e368d5e8015449daac5dacd0ceceecd1 : SUCCESS in 1h 05m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/28626a3ef7314de08c546e7f396a9ee7 : SUCCESS in 1h 24m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c24a2b37786748959a639fa7aec1000f : SUCCESS in 2h 36m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bb680336c9b24cf8866498f664d32dea : SUCCESS in 34m 30s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/2735da92c6ae4c2d91591bc12999c8c2 : SUCCESS in 16m 07s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/414d755d069b41589fe5f2be8c2cc4a9 : SUCCESS in 5m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2f8843188c8143a38f334d1d26afca6d : SUCCESS in 1h 22m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b085d38c160668fe5b2aeb63fe6f3b330392a83e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-29 07:52:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec3c78f57f37c41dadabd6f68c5ff8b903b9552f","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-29 14:02:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c5b67291b7735ec4f0795d3eae2a4ab4c8c5a17d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-29 14:54:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"854a10c5c349dd9e04cb45833360c8b0ba0db75a","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-29 14:57:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb26d65d3030eba6b58faae19d69758752bcbf2b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-30 06:53:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2df1f904b62d1e0790f416ad04626db285396230","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-30 13:54:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"dcab9921240d786ba6b9013ac560663f4194f56f","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-30 13:55:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d1e79e3b5726e2b4f786b363e4b91741c0ae32f8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-30 14:08:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f791b6ddac5b255d7ca45b8c447d39d46717b83e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-30 15:24:14.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"60b0c00dfce270cb53f30f435f4af26018e30053","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 17:18:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e4ed3de42434fd885a1043816442312\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/adb9d5746e6a4f36b6fa71f6791dbab8 : FAILURE in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a6a9a51a32c406fbc4d18aa7f08dcf7 : SUCCESS in 7m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f461104548814697830cfe3113a70d9d : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/264b795a4fa344a38ff7a7868ff6fe47 : SUCCESS in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b8eb585135d467cb4b28d348c1bb778 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a7260a240ed4e468f303ffd1629f775 : SUCCESS in 5m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a567995000b94eb388d2f7a383099ba5 : SUCCESS in 6m 57s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1524bb0c446b4f72b4c23d2634a0406c : SUCCESS in 1h 39m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d1d705da1e0a498a9b03018bc15cb88e : SUCCESS in 1h 18m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/87fe6b9a405e42a9af6d618251278ea0 : SUCCESS in 1h 49m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6461e9c558e849d287677498de0b62a5 : SUCCESS in 35m 20s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/e69617283e1f43c4a5632643ef0e4583 : SUCCESS in 24m 30s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/83c4a2bad210440892661566192b8105 : SUCCESS in 5m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4203f6a2ba8a431181637b8a655f1a5d : SUCCESS in 51m 45s","accounts_in_message":[],"_revision_number":3},{"id":"3fde833f9ca2e32636ba80dbbd4c38a7109133a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-30 18:57:08.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"05e4a196cc5f124459a048c668339252bf4bffd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 21:10:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ce2acf0e6c14a7b8d24e1360ebfdcc8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/62ceb756f95d4c13a610cf552f17ff55 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d136e19a9d8c479fa24d324d70767f8f : SUCCESS in 8m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49ec4d52c0eb4bd79bc438b8ba50fa92 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/238c9de0c95449dc91fd580efa56c8d3 : SUCCESS in 7m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be26f444fdb449c99955e1e2b0677774 : SUCCESS in 2m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b28c7c0f001145389b321bebaaaa28e4 : SUCCESS in 3m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ed212e92ac11467eb2996f3bd087c91c : SUCCESS in 5m 03s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5da5bb286f0440e68c706592010890c8 : SUCCESS in 1h 13m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/39803a91ef184b92903614770fbb24ac : SUCCESS in 1h 23m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/af34ed87e0544843ae89ffeb32c557ed : SUCCESS in 2h 12m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/87e0efdf0d894811bfa7fdd159d656ff : SUCCESS in 32m 41s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/cdff2a69253b4344a09245633bee4420 : SUCCESS in 28m 21s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d022f6f07d684ca4be70fcf936fcba14 : SUCCESS in 5m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f96098ef3b16492c9904726d6ef14bcd : SUCCESS in 43m 13s","accounts_in_message":[],"_revision_number":4},{"id":"b7ffe34f86fc33d0cc8266fe7e567bc35f22d89d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-01 13:54:22.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"faffce9731c736c1513ee9f9afa903eb6000d7e6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-01 22:36:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"622bb5aa8a3c03d56505d1d3b45c2af636e7ad5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 22:37:09.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9216804b73e8e4b37e3acf7b98df501077f18dbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 01:05:43.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/296a627a17514148874c5f4c7ba56ab6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f59a2a5b6ab5469da7dfe3b17cc403af : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b2c056c5504eee9e60f60e78f39bf3 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2f807ceaac246efbc32dd69cad6b456 : SUCCESS in 7m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f702ac18aa1c400c9e7435e69784eae2 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ec7ad91a144485bbd6f645481791fa3 : SUCCESS in 3m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21856416656b44cfbbb7af4b2ec71e5e : SUCCESS in 1h 23m 13s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ba0671a838474cf0919e1383c0dbdd94 : SUCCESS in 1h 34m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/479e21367776439b8c83efbb95773b39 : SUCCESS in 2h 28m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5a6a90d78cba431cbfb3f09e149750fc : SUCCESS in 1h 19m 50s","accounts_in_message":[],"_revision_number":4},{"id":"a8bec8b681f6ca86b40236540c764d59de0cb501","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 01:05:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"0cc2f1647011b2cf48e287e35cb1a7daa9c63ab3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 01:06:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/169b98658d2242d9a223fdbd28703937\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/19d6cf0ea0384cef8130ece550528ea9 : SUCCESS in 43s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b06cc50a3da64a2b9b83317895ea4f69 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2d5b2f39271e2953c3e8447a1806ae6cca98985c","revisions":{"da830e15004bf39d91607110478ce483cb92e46f":{"kind":"REWORK","_number":1,"created":"2026-04-28 20:39:16.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/79/986579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/79/986579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/986579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/986579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/986579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/79/986579/1"}}},"commit":{"parents":[{"commit":"37721a879f2a6ed6f5184487310f1ceb28991afc","subject":"Merge \"Reduce DB retry decorator retries to 1 in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/37721a879f2a6ed6f5184487310f1ceb28991afc"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-04-28 17:43:40.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-04-28 20:39:06.000000000","tz":0},"subject":"evpn: Add EVPN API definitions","message":"evpn: Add EVPN API definitions\n\nAdd evpn_vni attribute extension to the router resource. As the L3\nextension add_router_interface method does not have any schema,\nthere was not change required to the API and the --advertise-as-host\nattribute will be taken on the plugin level.\n\nRelated-Bug: #2144617\nAssisted-By: claude-sonnet-4.6\nChange-Id: I9755da5ff7912b35e78f782a6d2edb8992279682\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/da830e15004bf39d91607110478ce483cb92e46f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/da830e15004bf39d91607110478ce483cb92e46f"}]},"branch":"refs/heads/master"},"bad3daf8f280eb57e03e1df29f6d568ad8b6ecee":{"kind":"REWORK","_number":2,"created":"2026-04-30 13:54:55.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/79/986579/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/79/986579/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/986579/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/986579/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/986579/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/79/986579/2"}}},"commit":{"parents":[{"commit":"37721a879f2a6ed6f5184487310f1ceb28991afc","subject":"Merge \"Reduce DB retry decorator retries to 1 in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/37721a879f2a6ed6f5184487310f1ceb28991afc"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-04-28 17:43:40.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-04-30 13:54:41.000000000","tz":0},"subject":"evpn: Add EVPN API definitions","message":"evpn: Add EVPN API definitions\n\nAdd evpn_vni attribute extension to the router resource. As the L3\nextension add_router_interface method does not have any schema,\nthere was not change required to the API and the --advertise-as-host\nattribute will be taken on the plugin level.\n\nRelated-Bug: #2144617\nAssisted-By: claude-opus-4.6\nChange-Id: I9755da5ff7912b35e78f782a6d2edb8992279682\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bad3daf8f280eb57e03e1df29f6d568ad8b6ecee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bad3daf8f280eb57e03e1df29f6d568ad8b6ecee"}]},"branch":"refs/heads/master"},"a0756597679c8312ad286bfa4205dfc06c05b9f9":{"kind":"REWORK","_number":3,"created":"2026-04-30 15:24:14.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/79/986579/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/79/986579/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/986579/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/986579/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/986579/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/79/986579/3"}}},"commit":{"parents":[{"commit":"37721a879f2a6ed6f5184487310f1ceb28991afc","subject":"Merge \"Reduce DB retry decorator retries to 1 in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/37721a879f2a6ed6f5184487310f1ceb28991afc"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-04-28 17:43:40.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-04-30 15:24:01.000000000","tz":0},"subject":"evpn: Add EVPN API definitions","message":"evpn: Add EVPN API definitions\n\nAdd evpn_vni attribute extension to the router resource. As the L3\nextension add_router_interface method does not have any schema,\nthere was not change required to the API and the --advertise-as-host\nattribute will be taken on the plugin level.\n\nRelated-Bug: #2144617\nAssisted-By: claude-opus-4.6\nChange-Id: I9755da5ff7912b35e78f782a6d2edb8992279682\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a0756597679c8312ad286bfa4205dfc06c05b9f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a0756597679c8312ad286bfa4205dfc06c05b9f9"}]},"branch":"refs/heads/master"},"2d5b2f39271e2953c3e8447a1806ae6cca98985c":{"kind":"REWORK","_number":4,"created":"2026-04-30 18:57:08.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/79/986579/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/79/986579/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/986579/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/986579/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/986579/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/79/986579/4"}}},"commit":{"parents":[{"commit":"37721a879f2a6ed6f5184487310f1ceb28991afc","subject":"Merge \"Reduce DB retry decorator retries to 1 in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/37721a879f2a6ed6f5184487310f1ceb28991afc"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-04-28 17:43:40.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-04-30 18:56:50.000000000","tz":0},"subject":"evpn: Add EVPN API definitions","message":"evpn: Add EVPN API definitions\n\nAdd evpn_vni attribute extension to the router resource. As the L3\nextension add_router_interface method does not have any schema,\nthere was not change required to the API and the --advertise-as-host\nattribute will be taken on the plugin level.\n\nRelated-Bug: #2144617\nAssisted-By: claude-opus-4.6\nChange-Id: I9755da5ff7912b35e78f782a6d2edb8992279682\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2d5b2f39271e2953c3e8447a1806ae6cca98985c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2d5b2f39271e2953c3e8447a1806ae6cca98985c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
