)]}'
{"id":"openstack%2Fneutron-lib~986742","triplet_id":"openstack%2Fneutron-lib~master~If39fea477987e6dd253705f4e53d828009ed8a8f","project":"openstack/neutron-lib","branch":"master","topic":"bug/2150714","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-05-05 01:03:48.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-05-05 01:03:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If39fea477987e6dd253705f4e53d828009ed8a8f","subject":"Remove fip64 API definition and documentation","status":"MERGED","created":"2026-04-30 07:39:36.000000000","updated":"2026-05-05 01:04:47.000000000","submitted":"2026-05-05 01:03:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986742-bug/2150714","meta_rev_id":"494e8a934bd6ac725c78b3a44702537e285cd96c","_number":986742,"virtual_id_number":986742,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-05 01:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-04-30 11:39:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-05-01 00:09:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-05-01 00:09:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-04-30 11:51:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-30 07:50:36.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-04-30 08:07:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-30 11:39:27.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-05-01 00:09:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"aaedc3cf97de8a82e9bfa4fbbe6d94bb1011729a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-30 07:39:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37cf9619e41f98efb9c35ecae71befdd2b735bfc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-04-30 07:50:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e2a0fa10c91abec2e39b6cdf1612a68955f76383","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 08:07:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad5240608aa1425c83e4455118308074\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e3d44f69c6374561a5a933ae90ebc478 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/939ef513a99d455fb2d4528a87d467e3 : SUCCESS in 7m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae1da90e681842f2aea18d8f0fd62288 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee47bc3b41ab45c095c45998a9e92728 : SUCCESS in 7m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec6c8a9ff7c847ad86d95d2aef539999 : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55e650ab74074ebd95418337cced6562 : SUCCESS in 3m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b4a342e9650b47448cf76cf54d1e0957 : SUCCESS in 6m 17s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75cb97f47c824ef49e8de594d719e4d1 : FAILURE in 27m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/355550ba4c7a4c898695a17985565cd4 : SUCCESS in 4m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d8328a3c7516415f84ce24f1c21f7234 : FAILURE in 15m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9e6ead0bb45a467e85bb79aa1c1220c6 : FAILURE in 16m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/32f5cf91c0ff4ca7af0019eab4025d09 : FAILURE in 15m 59s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/76773c80248b4dbab42c02e514b2b2e0 : FAILURE in 4m 17s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e5468e68e6ae48f7b5d9ef33dbd6b8b7 : SUCCESS in 4m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a75add254c11413c9caf039371f2aa92 : FAILURE in 21m 22s","accounts_in_message":[],"_revision_number":1},{"id":"21870ba8ddd8f31d549a0107d42f7d96a230849c","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-30 09:36:05.000000000","message":"Removed Code-Review+2 by \u003cGERRIT_ACCOUNT_11975\u003e\n","accounts_in_message":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"_revision_number":1},{"id":"9a148ee3ebaa76e5460921954da710f60be7f5bc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-30 09:36:33.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc85a6d4e7fa3b3de6cdca62f4ca0b282dffeb49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-30 09:48:57.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"505ca96f3d7d6a0468e90719b4ab526833b0b04e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 11:30:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e83143b3c7ac4db9ac2f4920f781c7aa\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a3eeac8ba32c49b692f7cdbe7bca9d14 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f0cd0b4037f4781af6e15f218888ce5 : SUCCESS in 8m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49722f4b06884c24bc96583e9f74e3d8 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61605248ce874d3c82536be5da127307 : SUCCESS in 6m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6788c769f564def8eeaedd809ef9d72 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/868683cbefe14ad6867b62122a003fbc : SUCCESS in 3m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3e8477b86dca4ef2ad4497c6a2d16bc8 : SUCCESS in 5m 16s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd4f3bd7777c428cb4a88dac61658b98 : SUCCESS in 1h 34m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5db62835589146e9b6d9ecc1590e2de0 : SUCCESS in 3m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b4203ea18daa467494b4817f669f9db1 : SUCCESS in 1h 25m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e31b9406cbd949b49c53e069e8296d71 : SUCCESS in 1h 38m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aa087e18cad04dc5a285bce37bdc5b17 : SUCCESS in 31m 46s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/9f259b36f82b41c0994ec211879900e7 : SUCCESS in 16m 13s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/fa3b675e23644a688f2c8f27c6c2711e : SUCCESS in 3m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/034733e9c6054821860ac5a91da15add : SUCCESS in 45m 07s","accounts_in_message":[],"_revision_number":2},{"id":"f2e85e4c722f014241d2540ea8caa3d4ca8751a0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-30 11:39:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9298a182bc9264feb81e4eb3a253da630db78315","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-30 11:51:33.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"0e7fe160ea40017ec94c59a44c5de31f81c79328","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-01 00:09:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cb0cb053c661a151f2191a55733643b0e085682e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 00:09:50.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0cc19050f464a83b33d1e623dc1d953\n\n\nWarning:\n  Change 986763,2 in project openstack/neutron does not share a change queue with 986742 in project openstack/neutron-lib","accounts_in_message":[],"_revision_number":2},{"id":"b7e9db3b1ea525a6b720c87bce185aedf4ec98f8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-04 17:00:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb34b160cf21e7aa8936a9eaebc5056409417d6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 19:33:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/869a67ebf34042ba8f947ea34e9126fc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bd8db56f35274e83b7cf61b0e54f7536 : SUCCESS in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98d6fc19f221462f8528aaf2f3e9117a : SUCCESS in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9eb6c95bbbd44fc885d65a448367098 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd3c2f0d72864bdca8e6eb020178ddb4 : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97a9eff98d354a5999c49fe99c3c2e5a : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/737c1b1065cb4887aafc9270cdbc3b58 : SUCCESS in 3m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/03af6d07c5824e11bab183c19e4ea5e3 : SUCCESS in 6m 05s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6bef4c334da4b578d487fb5a0e4ad0d : SUCCESS in 1h 39m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5485331d7e514e0ab506decd60060d8d : SUCCESS in 3m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e89ce50caefb414f9a968d0f76b4cd56 : FAILURE in 2h 30m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bb343839e9954be89e474f75881604de : SUCCESS in 1h 53m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4bffa65e6ce14e7486a9d1a1a9caa89e : SUCCESS in 32m 35s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/2b20c90c903a44f4aafafbbe9bd3b559 : SUCCESS in 26m 39s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7523989539a74199b91c6f57dfdb3123 : SUCCESS in 5m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dc1411097eea460f8182671beba2382d : SUCCESS in 48m 26s","accounts_in_message":[],"_revision_number":2},{"id":"4fd149df93ad9733416f1025186226c29a205c43","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-04 20:56:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ee82b3ae62796d656bbca87d709d614cb9b5a496","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 22:53:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e5dfb6344524e39a3a106da5fb27207\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e9728e347b6a481397f8d7a263efb014 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20922799b9604b6e96552cd4102ea276 : SUCCESS in 6m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c8a376a51d149bdbc3a998d7a3bd704 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d5bd1e7b0e64cd5a87ca23faeb9a8d4 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b1e7e2f90e44d59b6f24cb672045ec9 : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b1bdec4055040169dad97413f013798 : SUCCESS in 4m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae733152b9f2471c98b8d667915dde2c : SUCCESS in 6m 02s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5371f98873f44a9d90f8164f1cad9957 : SUCCESS in 1h 38m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1cade8ac09c4d9a97579ef1359f16d2 : SUCCESS in 3m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/56aabe35170b4f639b233b24226e9e1d : SUCCESS in 1h 20m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dc2936b359ed4887963bbd143c5dd5cb : SUCCESS in 1h 53m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/31bb7c117deb42f489e21434a9fc317c : SUCCESS in 32m 37s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/a3be47665ef94889865827a87631665b : SUCCESS in 24m 12s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f197140c1e8a473b92d9b3a78a73808f : SUCCESS in 5m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9f7137ea04ac4aeb8318654b71fd3c54 : SUCCESS in 1h 25m 52s","accounts_in_message":[],"_revision_number":2},{"id":"3a016fddfaf8d15287bd5f7d32fcb5d72197ef86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 22:54:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"878b7d1007e23cda07f04088b8804648e2def560","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 01:03:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaff90a0556c48e8806307c84fdf5168\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/568f1e58f69044b7894d8ff60eacb188 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5e1e53020f3426fbf6563b9ae2eab31 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d65c69bec228492ab1d3e1de78b16f33 : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f771a2fa73234cbe86cf9c99f1ba811b : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f93c28cc73cb49ffbb9899ed6586f7dd : SUCCESS in 3m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/549ca406750145a4a4852978980b81e9 : SUCCESS in 1h 47m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48b9763d7c324f4888dd166d9ef403b3 : SUCCESS in 3m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/71d2686a94044a279aa3c27784be28d8 : SUCCESS in 1h 27m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8e1aafd5c4d64551a925f7e76adf4654 : SUCCESS in 2h 08m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a5e5ce508ba04b8595e1629a82cadfeb : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":2},{"id":"85c07f98293caba7c72bc23e7cf3580467e65ada","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 01:03:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"494e8a934bd6ac725c78b3a44702537e285cd96c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 01:04:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec267e92c83c4998826a2461c7a975ca\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/ec1d6d15795a483b92d6e24838e973a9 : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9c1ad69257143f3900f532a7c1406a0 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08ae6231e4174bd696be8fde066c1270 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"da581b7bd321d4706d6c6db90928b78efc945858","revisions":{"7a7fa294f6243ea3121f5b35bc305847ee7f8efb":{"kind":"REWORK","_number":1,"created":"2026-04-30 07:39:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/42/986742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/42/986742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/986742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/986742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/986742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/42/986742/1"}}},"commit":{"parents":[{"commit":"fe86c1988a2fed008de27b055d0fd127daffaf2a","subject":"Remove logging-resource API definition and documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fe86c1988a2fed008de27b055d0fd127daffaf2a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-30 07:38:57.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-30 07:38:57.000000000","tz":120},"subject":"Remove fip64 API definition and documentation","message":"Remove fip64 API definition and documentation\n\nThe ``fip64`` extension was only used by networking-midonet, which is\nno longer maintained. Remove the API definition, its unit test, and\nthe api-ref documentation.\n\nCloses-Bug: #2150714\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: If39fea477987e6dd253705f4e53d828009ed8a8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7a7fa294f6243ea3121f5b35bc305847ee7f8efb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7a7fa294f6243ea3121f5b35bc305847ee7f8efb"}]},"branch":"refs/heads/master"},"da581b7bd321d4706d6c6db90928b78efc945858":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-04-30 09:48:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/42/986742/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/42/986742/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/986742/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/986742/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/42/986742/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/42/986742/2"}}},"commit":{"parents":[{"commit":"fe86c1988a2fed008de27b055d0fd127daffaf2a","subject":"Remove logging-resource API definition and documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fe86c1988a2fed008de27b055d0fd127daffaf2a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-30 07:38:57.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-04-30 09:48:56.000000000","tz":0},"subject":"Remove fip64 API definition and documentation","message":"Remove fip64 API definition and documentation\n\nThe ``fip64`` extension was only used by networking-midonet, which is\nno longer maintained. Remove the API definition, its unit test, and\nthe api-ref documentation.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/986763\n\nCloses-Bug: #2150714\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: If39fea477987e6dd253705f4e53d828009ed8a8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/da581b7bd321d4706d6c6db90928b78efc945858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/da581b7bd321d4706d6c6db90928b78efc945858"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
