)]}'
{"id":"openstack%2Fneutron-lib~986960","triplet_id":"openstack%2Fneutron-lib~master~I74b11c43aedec829e16296a3d074c0b38cd5728d","project":"openstack/neutron-lib","branch":"master","topic":"neutron-project-id-migration","attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-08-10 21:25:34.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I74b11c43aedec829e16296a3d074c0b38cd5728d","subject":"WIP: Tenant_id has left the building, part 1","status":"NEW","created":"2026-05-01 21:43:37.000000000","updated":"2026-08-10 23:28:29.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5cf81470d5ad4c3ce68246f80fe409c48ad2ecda","_number":986960,"virtual_id_number":986960,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-10 23:28:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-01 22:18:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-10 08:37:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"12371bef42d864cd244ef4da9ac7e1ef6f4a8692","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-01 21:43:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cd99acbfbd205b40baf214a1d5868222027104c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 22:18:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71281561141849c58468a4ae0265d34f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/744e0245baef429384e2ee70c92be1be : SUCCESS in 7m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74208b72ee4a444f9435ca68d1a7534a : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e51981da75154db7a6be712f21944b69 : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ddc7c3f5235b4038ac70dabf0b8341e9 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6cd0f58a401242b8aee31d5412fb785d : SUCCESS in 2m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a19b23b072a248b0a9739c41351b24f1 : SUCCESS in 3m 32s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e98ced5ea9414de2a5c2d214d965002c : FAILURE in 29m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0136e9b18e444cc69be67e99cd043892 : FAILURE in 14m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/78470d8addc4437e80d11537a159417f : FAILURE in 14m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ee084b5756eb4ea6b39c2389b56eac73 : FAILURE in 14m 57s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/6d1c4c9729c248e683ef7cace714fcc6 : FAILURE in 23m 58s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e066a7d0d773453f8f42a1005b98dd4a : SUCCESS in 3m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6a044983dc35422faf33f63886088278 : FAILURE in 28m 10s","accounts_in_message":[],"_revision_number":1},{"id":"0ef03f4a8d813b987a549208d4071d474553fb72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-01 22:20:18.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"e5390a3579b71321dbeb737d86c4abc3db8548a8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-01 22:43:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"599f6b79edabc362efe59e52293f290923a8c124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 23:02:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51d8473f6b6341e8b622c91c92ced7e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cf359eb642045d39af6755e22b990f9 : SUCCESS in 6m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/265cb1e93a5b410cab566cd34201133c : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b4e7ada9c784bc79898a4720b85b50c : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc0488c78b5b413c996256fcdd8951c0 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3ac8652f18a1430b9501aafc7c8a7b06 : SUCCESS in 3m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e2c8850479ec463190bef73e46c70c1d : SUCCESS in 5m 16s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad1cd56ae6a64879af0cfcb34e3b2445 : FAILURE in 15m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c83f370a634646b696270fd2d409eded : FAILURE in 13m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/86a2595315064f5aa098995a6f498c5b : FAILURE in 14m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2c0d0888563a486093b214cb8f40a2ee : FAILURE in 15m 11s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/1b1b8f048ec7465db1543b06471a099b : FAILURE in 28m 50s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/cdf74e3fec6e4e24a98c6438617709c2 : SUCCESS in 4m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9f3b3ed67dff48488ae67ea4fb82289a : FAILURE in 39m 29s","accounts_in_message":[],"_revision_number":2},{"id":"dc021d933c122a2d7c7da681b8109426f0f7350a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-02 01:39:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ccf06b09f6f624d8fb371cd514c984375a1356e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 02:29:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9dce9f9fa894571a298c76757a39d91\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f69e8f81594762ac520a7f0f3312a0 : SUCCESS in 8m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2b7314388f643ccaa06818f96c4e72a : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03e8b061643747308b1092a85870f378 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15559369156f4b90a23e6d9fdfb066a9 : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/583f27a72c3b44eb8b819d7d253dfcb6 : SUCCESS in 2m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5aa5a73af16743b79fefbcbd4bb968d8 : SUCCESS in 5m 57s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6646d475f56b461486ae935f230a7beb : FAILURE in 24m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f2122fcf93774d0c948645edade37e9b : FAILURE in 13m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9583c4e820d847cb98584c4ede14957b : FAILURE in 14m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6ed78b0e43e649beb33f5d841980ca1b : FAILURE in 15m 20s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/168adb7cc74f428c8da1ef5f9aab7653 : FAILURE in 24m 01s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/bd740db27919439fb7c1714f709bfb3f : SUCCESS in 2m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3c7956ec2cc34639be0daecc0a7d6364 : FAILURE in 42m 47s","accounts_in_message":[],"_revision_number":2},{"id":"ad880eeb7c4f0819fbdb2309069b96f0f6b0b102","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-02 02:42:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"55af94fd57e26d0a8d065aeb75380f04c91c71c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 03:49:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bcfe02587b540c5b3a78fb56af66a12\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43b26da3cc8f4a8bb0755e20ac5500b9 : SUCCESS in 9m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/044ee072d59d41b7bf639c6647562882 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c26a2435b4a64075a5b78b4ef8f66206 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/428ad9bb28c1475e938e297867ddcdea : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a2ee011eba44c6b99805d96a299cd62 : SUCCESS in 2m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0c4c037ccca84c25b3bd003b7f3ba8d2 : SUCCESS in 5m 13s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fae141dbec4a4626b9aa79df50669c3d : FAILURE in 26m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dbd4ce463d5e4f8dbbd378af696465c3 : FAILURE in 17m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fa35265814cf4ebe85ff37358821c663 : FAILURE in 19m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c4b2b4c90e9a4d9b97766ebdd19500bd : FAILURE in 19m 24s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/69fa969f2ad841f9b2ed7179feb1fe70 : FAILURE in 26m 37s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/22c5715b6bf54fd39c1fc53367fa1316 : SUCCESS in 3m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e9e0e1444cea40c8af93cc91c65103ce : FAILURE in 44m 21s","accounts_in_message":[],"_revision_number":2},{"id":"eb9400e9dba22dc28a9d8026118f15ad9a02ee7f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-05 20:38:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"54ced9b59c56fafa9f551a8dbbe5c78a656b3f5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 21:09:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/438a168c6c514a32a9e5c0f163969d41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e32b6a953ac4774a5e45c8dbbc35c82 : SUCCESS in 8m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61aecf6499334c34a65afb9b9ede9c3c : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c320faf462b4e9db16a222ec4ab9474 : SUCCESS in 7m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c53c6c3881b749ba9485dbdd6acb37a8 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e18e72a5f22a49f2b6751b709fdacdfa : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/994000addeba4d3e85d46519abdd086b : SUCCESS in 6m 09s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ecbed0a409bb460ab8778a4152d64c60 : FAILURE in 14m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3dd918423dce49509580427ef6b7e23e : FAILURE in 16m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/263d1473e10541688ef2f87dfd8c5e2a : FAILURE in 16m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2a65d463ec7d4f26bf45a95823abb97f : FAILURE in 17m 14s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/769c18dd0b904eaf94dcf5d71be3ed13 : FAILURE in 14m 59s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/39ff4992220b44289bbc0d013cfc17d2 : SUCCESS in 6m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2328159e783349c28b456a8eb7fc18b6 : FAILURE in 27m 11s","accounts_in_message":[],"_revision_number":2},{"id":"3fd5e09748be132c23ae36e82929d1a69df716d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-06 05:03:37.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f14bddeeef4a5d5f794746d1f318e72ef896bfa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 05:30:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cce5bad04164578a3b43565bb662568\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/461593782bee4c3bb6dce2f1a9b2df78 : SUCCESS in 7m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1329fc58db37488e8b38464a539744b4 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9982e46d60242a29086886c99ce301a : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/68636c1899904c87b4df3c2d41f472b0 : SUCCESS in 6m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68bf901a54e4437dbbb88bfffb57ba34 : SUCCESS in 3m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/02ca86cd62534ecb98b2f3cf0b2303ce : SUCCESS in 5m 39s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1071fa59893c4e2e9919727ed27e0727 : FAILURE in 22m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7cac7055769a4534a8793949c67b6b47 : FAILURE in 15m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c5bd70abd8984f80849f51fb20810064 : FAILURE in 16m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/37438512a0634c8585c1a2064a39139e : FAILURE in 15m 11s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/1c3250ae03e24cb18ada2ac351b30008 : FAILURE in 22m 41s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ec0fbb5805034d6b86b73dbdced3ae42 : SUCCESS in 5m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1d968b2e8bec4302a5025188160a9653 : FAILURE in 25m 03s","accounts_in_message":[],"_revision_number":3},{"id":"86820aea2b0747c8c9ffd6e2bc152d8252bb5477","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-06 05:39:35.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"aa909fe0cefe44ebd44f0d4b40e0bc3c31184efe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 06:25:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59422b85f65f46baaad09254bd6c2eeb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c29055cc6894b589a8e4d9e8c1d848e : SUCCESS in 8m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48e8ef5a5d24428ba2d950726a345d82 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f82eb49398a64c4b82294bd6585e3907 : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/730ccb79c12645909c399d5ef43202d8 : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/495702aa5fe044eab26f2800ec30e450 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0c8e9f3b98a94efd83c633cca684aff3 : SUCCESS in 6m 20s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02eba626a5df4c39aa2c1c74badec2e8 : FAILURE in 10m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/393bce16a55445d2a46b2c224f44286f : FAILURE in 14m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/582217c1ba774bf4823e05fc5e2cce8f : FAILURE in 15m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c772fc6c21f74d079fd3cd31bb9d2117 : FAILURE in 15m 37s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/fe22f67c31b645f9849b7522017166e0 : FAILURE in 18m 42s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/69ff8c6900144b488258755b25f9d152 : SUCCESS in 5m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9117be9ef3054170a087d7cd72a62d55 : FAILURE in 40m 45s","accounts_in_message":[],"_revision_number":4},{"id":"7789b9b33a4095a8889b86b8d9cdf95b043b4a70","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-06 19:16:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cda9a9bc3f092caf0962e726bb3d6e746babb309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 20:10:26.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9bc34909244d4a188977893aabf16a4d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa404a78082a47f1ad772f37869d612a : SUCCESS in 8m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34e06cd467e74487be52b5896d0c1749 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99cd907a65404cb28af0b58823b2184a : SUCCESS in 6m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34d262d72c9747dcb247e67a341bc707 : SUCCESS in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8219fb10b734e23a8a39c31ac251d46 : SUCCESS in 5m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/46cd65e24d694e488468afc2a59429c5 : SUCCESS in 5m 51s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5578dc267094a189d11cc382b81cb83 : FAILURE in 27m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f9b2449452b64e45aec2e40aa249df2c : FAILURE in 14m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c4050c099d24436a915bb6e571d42b9c : FAILURE in 15m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a8af443f0a424dc6b19e0f020caee6d5 : FAILURE in 16m 37s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/25fb11811d894696b7ad70bff1cd9f42 : FAILURE in 20m 36s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/105004f9b2784e91942d0651b81ef4e2 : SUCCESS in 4m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1798e7b769864b628beffdd701c63364 : FAILURE in 48m 17s","accounts_in_message":[],"_revision_number":4},{"id":"5113967f0d6eb9283551fb577c7d4ce9a1e8e27f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-07 14:20:48.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"b93c7ede665f535d28c45b725ca64615cd8e5145","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 15:08:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/489011b356d34c8296a64c7938228828\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/942d157899814afbbc632ace25e34e37 : SUCCESS in 8m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/572c8e71115b4db59f4fa8fbf4bbf49f : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79d64c279fe7411dbfed31470fa5779b : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/90269866b3d5466884db5230f7e19eb8 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a7ff04dc9534817b9a7cd0f063264c2 : SUCCESS in 3m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3d3f6935662f43f2bc234c6348d68bf3 : SUCCESS in 5m 38s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b249a4e4bf1b4bedb1cd6a713ccef5d5 : FAILURE in 20m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6be440613b2b4d21aa653b3a71451791 : FAILURE in 16m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6509bb8a86e2491bbfabd658605aa0fd : FAILURE in 16m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6fc68c8944fe494ab2d004f3a21788f0 : FAILURE in 15m 05s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/598882a0b00b4ee395da7700bebf0353 : FAILURE in 23m 02s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/161814a7304f43d3ba16557fb312c127 : SUCCESS in 4m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/31a7ec25a63f4bdbaf3cd8611d2b28c4 : FAILURE in 40m 47s","accounts_in_message":[],"_revision_number":5},{"id":"505ac2c556883408e13aa1ec2eb62dcc398d2558","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-07 19:34:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"43b7dd075142075e7d3b9d84a77dd40ee85d91fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 20:12:34.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87664369ceed4c869c85daf318431aab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d4feb29a6e8401182cb19de8b8bed4e : SUCCESS in 4m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/135de1f6345148dd93b49a966abea384 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57e226ee935c4019a9c809e3b6b561ba : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7aeae08d1a4745e7a3f19dfe2116576a : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29492c44c6c34ea5934bdec792d29f2f : SUCCESS in 3m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b132bb4d1ae34d4a9c7ca971e5d45a9a : SUCCESS in 6m 23s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32ca3a03d12241149890b0f69a1bc402 : FAILURE in 24m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7fa849bbb75a4c9faeb050c228a1e62e : FAILURE in 17m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c002b1173cac4392a08ecb379fe71a10 : FAILURE in 18m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/808e30438461433ea72ccc8ef48cc403 : FAILURE in 19m 09s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/14a2b1ebaf554fb9ba2c7325c401e473 : FAILURE in 20m 08s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/05c4038dec9c453aa74077f14755c10e : SUCCESS in 5m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a73d604b89b7476695438dc07a7dfdab : FAILURE in 33m 48s","accounts_in_message":[],"_revision_number":5},{"id":"0e8e6dad276722399851ffc7f25abec9367661d0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-07 21:40:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"adfff687ff567595104c0fc178815bb56b211d81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 22:42:34.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c497a7533af4d0fb7056103f0153074\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dc430b15b134ddd847ba42390280615 : SUCCESS in 6m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c90e99e0a5274f9d978d3876e841a6ec : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a4161d2ee6c4781aaa9aacb06b47679 : SUCCESS in 7m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/49c028ed791e45bab071f6410205c164 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1c691f1cab6a4a19ad9b5c230f9b15cf : SUCCESS in 3m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/144323cbfdae4f6abfa99f3e396d5282 : SUCCESS in 3m 48s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9920e9295b743c0808af6ac09032f49 : FAILURE in 55m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5912a10d7d2c49cbb26ce32cf7c8ade2 : FAILURE in 38m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0ae334c13be04b5b807f551b0b7a5382 : FAILURE in 41m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6d90d9e5dfae414da58634ec91c0ae9c : FAILURE in 31m 23s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/fa87eb0a60fa40e08b1b4207b326ba15 : FAILURE in 26m 27s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6a9b63980e94498288f4607ae1fc395d : SUCCESS in 3m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/15a3c066320c467688f5dbf8d5b9d1a2 : FAILURE in 26m 56s","accounts_in_message":[],"_revision_number":5},{"id":"a6a98a137c35cbbdc5cf72f70635b097f8120134","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-07 22:45:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"918511ab789e233c2e0671677124148107543c63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 23:37:49.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45fff98f004e4338ba9e93deb25c740d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c733a1ab4ff44c85ba436a663ed9dcc6 : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c3ebd7155304527b79e3c8ea72de4d6 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a02f7e9950054a07918db596bdd3c016 : SUCCESS in 7m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45a5dbbc47ae463587c65ec949567bd3 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7622381e210e4a3598c7d907efd5bc5a : SUCCESS in 4m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2e3cc65b49e3463cb3b56234d172c9d3 : SUCCESS in 5m 28s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d93b2fb17ff4e91b2ba1dfc41ba69f9 : FAILURE in 38m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c9f9dd339f6940519b058c7186c1eb16 : FAILURE in 35m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5e18e4b1606b4a96948a1a424b93cca9 : FAILURE in 39m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2e9733b971f14ffbb181574cfc2fa4c9 : FAILURE in 30m 27s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/198cdb0cbce14e18bd536357fc9d4fa1 : FAILURE in 22m 01s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5f8dd30cb05b40909304b0b9d20bcb42 : SUCCESS in 4m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1b6dab4435634fc8983c89ba2c7de28a : FAILURE in 46m 45s","accounts_in_message":[],"_revision_number":5},{"id":"e3d1d459a046b95fb2a49754625ee079a26485b1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-08 02:24:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"55bab20486f35f6f66fcf59fa800c26ad6fcc8e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 03:39:35.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac955d86a51741c3b53d80f7dfcb3419\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da9ac5015a544071aa9e153ceccdd9c5 : SUCCESS in 10m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71359af8336a494096953329cf707fbd : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aa95bdcb2534b48add69f68c04a748e : SUCCESS in 7m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/227c1d8b3bd84a3b83de1b350c5ffaab : SUCCESS in 5m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/074f5715b8bc4375966fdd0ea0c61ce8 : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5cb1cbd5d5914ad58a100d96dfeba762 : SUCCESS in 8m 10s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/257a23b43a464118b81366cae8cceab1 : FAILURE in 44m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/18184de4b3e841fab44d3a7db22cd585 : FAILURE in 38m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ccaa15134cc843b78f921494f67d39f1 : FAILURE in 41m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bffb5582ee5c4cd1979d02486a444788 : FAILURE in 32m 45s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/1a5a83b7edb447deaec02158b2a4f6e2 : FAILURE in 20m 56s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/067ae174513d4838acb045eb601c3c49 : SUCCESS in 6m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f1a8a60803d24ec3baf66a407bcb1fae : FAILURE in 44m 47s","accounts_in_message":[],"_revision_number":5},{"id":"7d1e4c2d5fc6760e40323b73d9d169c91b491418","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-08 04:30:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"47ab4426800398daf9ed8caacdecb565d241e8e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 05:16:46.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a9d0fe7972e47058ccafddb9376daf5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10b5689f66974d15a2e62cd4d3f6aaa2 : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21a9f6c08c814e198022279770924ff8 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ebc7e86586d48688327409d6368b233 : SUCCESS in 6m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/867453c89e31480e9161ab48087adea1 : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/508d3edc2ced463f9dc8fcb710a01099 : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dc54a4471afa4b748634914148bfbe31 : SUCCESS in 6m 21s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62ee304468344088bdb4864e51051a1e : FAILURE in 36m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e1b16d9b721648f4b784b4927fc90acd : FAILURE in 40m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/db6c8a8115614b5e9429211a7ec3ba79 : FAILURE in 40m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6cddd2e5a8844f0db1c4c273032d5197 : FAILURE in 32m 23s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/822d2915118d43cb923ab87f511ab6ea : FAILURE in 24m 35s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/b931de9a5c0d4682880eb41735a85850 : SUCCESS in 4m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d3240a6afca1487aaf94ade2a61f7007 : FAILURE in 38m 16s","accounts_in_message":[],"_revision_number":5},{"id":"f09da91d75fc1c86f7e20b0aafe2907bc03d1255","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-10 12:37:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e24305ae8ab93938a6ecc56a767d5677706301c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-10 13:37:48.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6648b3eea1f941a787932cec26f78649\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e458c511656c4f808f8c083defea1811 : SUCCESS in 6m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/572dfd9fbf104e5283189da4b312dab2 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3948de3ea8154597afac29731b60148b : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a54b4b8640c24007a2c013b3c168ccb9 : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/35798989356248f6b17369bafd4235a2 : SUCCESS in 3m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6da8dffbf00942699e4999228902274a : SUCCESS in 3m 27s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e5bbdae09374e438a0fe778b6133f4d : FAILURE in 48m 32s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f6727ff52e424ca88e74c471fc8f8968 : FAILURE in 27m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/486fa3b3f2ed4da08330709eb41fdeaf : FAILURE in 28m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c6ecf604553c45f388d796234795d362 : FAILURE in 28m 42s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/f282114afc2a4978b6a6a193530203ad : FAILURE in 16m 22s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e45567db8d704f738993436cdb7c71ae : SUCCESS in 3m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f6049b55f17c4dff8983b3cd0b25942a : FAILURE in 43m 10s","accounts_in_message":[],"_revision_number":5},{"id":"28430c6aa4a00ccfc8c76ee5235a2bc5f6007fe5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-18 21:57:46.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"49b28646b40cb476dbdc86cfa04dd0b03c09b50a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 23:00:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a342286dea641eab5fa3192429007e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/001725ab9733432eb00aac385ec2df1a : SUCCESS in 9m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91fd36b4b28a44f8a460391ab0c4adff : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69da59513d984d8dbec081c615ab25d7 : SUCCESS in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/713dcd04ae70435e8480eec036bff9f2 : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8e3ecd3c0944a87a847bed13db311e9 : SUCCESS in 5m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1df8158ba0824008b1a49de5ef1865cc : SUCCESS in 4m 04s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5bddce76e91d4f05b7807d323f48447a : FAILURE in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33dd65c795674937b61c5af5f19452b9 : SUCCESS in 3m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/871057f150244acf9cdae7ede30f744f : FAILURE in 14m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1b0f79c829734b7e98a8c6493485410b : FAILURE in 15m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0c1d1a0f4a9a4d218494f1be8c3c1a6e : FAILURE in 16m 02s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/97b8267f094748948c0869439eebba80 : FAILURE in 18m 50s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ae05658f4aa249f0b106ce233efc7c7a : SUCCESS in 7m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/841eecd36c9c4aad99fc125c47f0fc22 : FAILURE in 47m 22s","accounts_in_message":[],"_revision_number":6},{"id":"aa6b6a935706948b48c6cf1f573abf89da85f4ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-01 00:18:39.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"43371548511a061ced14bc8892138f90b30602ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 00:48:36.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6179238e545041acb085cd0bcc506bd7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4995a9b296a447397078294dccc3cc3 : SUCCESS in 8m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e6f9456754c47a3bfbaa42d2ab00e22 : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/459bda2057324a99a3e03653c2acde0e : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb6a7874842a49b8be649a722fbacf6a : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae6d5438350b43adb09e241e3e3b0fa2 : SUCCESS in 2m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/db538cc12a5f4dda95d0c59c74df5cb6 : SUCCESS in 7m 27s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f27ce294666c47caad7890415b270c74 : FAILURE in 23m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1bdb6736ba44cbbbe963c77b7428b7c : SUCCESS in 2m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/211fdd24a0da44bb88950fdb287bdeb7 : FAILURE in 13m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a238231f8f1c436eaf3822272e529b34 : FAILURE in 14m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/92e66927e2e841e3a0a838cc1c01df7e : FAILURE in 14m 47s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/d773d6ce66414abe983ebad626f38c9a : FAILURE in 23m 04s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c998007d2b5c4f3794d6aaa2de61af89 : SUCCESS in 5m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/714672751cc542079e375cf48f846c88 : FAILURE in 23m 23s","accounts_in_message":[],"_revision_number":7},{"id":"4cc58bf8f57fd26bd5b2a0b365602d9a827e5d19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-01 03:42:32.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"3da4f326fe4d18507a5d11310047dc0cbc6e3695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 04:17:28.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84715ff83e3a465caa299ffcacb07b4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aa5b720c14846888427c896530777fc : SUCCESS in 4m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cf7478478844fb3851a4e733085cd83 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc75af65e6ca4bce9f60f458d04b6375 : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b940a7757c174a46bedbc7ae0b1a9052 : SUCCESS in 2m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c3a1e57a8e78491192980a3894f614d4 : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bc20d13ca5e643b1a86ce65fcc500c02 : SUCCESS in 3m 24s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3281325e37f49f3ac377506d037e20f : FAILURE in 26m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72a3e15cf17f4a178f5595cd5732910a : SUCCESS in 2m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b27643bb82c84993bc1bc919476a616a : FAILURE in 13m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f1e7f36c87b14c6c85a8f696fdff01fd : FAILURE in 14m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9377f43d728f41b5bdd5439abf5e5f84 : FAILURE in 14m 44s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/0ba161f3c72845b594d01c5abd85a7e3 : FAILURE in 20m 37s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2f24e1d0dbaf4af39d7af34f5dde9909 : SUCCESS in 3m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6ce53e37f2704a968e87122afb0c4bf0 : FAILURE in 27m 10s","accounts_in_message":[],"_revision_number":8},{"id":"f473a825be2d0a19252b9340dce9e7e233d84fb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-01 17:15:06.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"b8417ebb1ca7908c5bdb90a178bb5fcc22740623","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 18:09:44.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9345477c21b74270a3c322be45b28213\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/934b6dc43a2c47dcaa70904b26cde91b : SUCCESS in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e68beb1c3e234e7ca65a08d3a9d03271 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9f3d63896904b82826822832e33aca6 : SUCCESS in 6m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94632e5e42254cc4ad17a05f48382640 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1158dae6d3594fd888b8c26c6db42161 : SUCCESS in 3m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e5d13dcb36724607b65cd3818fbbdbdc : SUCCESS in 6m 05s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/066e5be54db3486d8017c0e87ebcc7d6 : FAILURE in 26m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d80e39ed94464eb6a2a809fb15b56d30 : SUCCESS in 4m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2c58fea1d60540f0912b0f18e50e15f6 : FAILURE in 13m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cd0f8826776d49f48f88653fabcc9eb3 : FAILURE in 14m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8555b75b514c4532aa51a2239631de80 : FAILURE in 15m 25s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/f7446e67063d45d88c3df566a13b6906 : FAILURE in 18m 58s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/25ebc980c80847b9903ee87237177770 : SUCCESS in 5m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/09cc300a59ad49cbbe6a3ea28f334af0 : FAILURE in 46m 31s","accounts_in_message":[],"_revision_number":9},{"id":"05b7154e01b247f3c389ea01be5658dbb334aa1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-02 02:08:47.000000000","message":"Uploaded patch set 10: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"d8c004087eabae3b1bfdb8ad772ba45d162c1e19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 04:03:57.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a14ee904489d436f868f07dfcaf2c219\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3097112dc3b342adbf7c700e077d14fb : SUCCESS in 5m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2394ee60685943a7b5e672f54f4c47db : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3a6c4a768b2437687582faf9836d991 : SUCCESS in 6m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23ecb239d84f43ada5787cdd2732c580 : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fff6f946ca1742d9923f4d32b1a91c97 : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4a266bbc0b6249cda1978b8b2d7dc9b8 : SUCCESS in 6m 42s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d38c5b945854b8388911bfcd020cad3 : FAILURE in 56m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/199f4c44db5a43c48afd6f11477b7f43 : SUCCESS in 5m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a9abe4deb1c441c78ba0c107cc8484f3 : FAILURE in 53m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6ba12045c7114ef9bf0281de7960e644 : FAILURE in 57m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e98eb67de7344229bdad10695798ffb2 : FAILURE in 31m 52s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/c32ac8587fd44b668fb1d0ee62c359db : FAILURE in 26m 30s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/dde141475f4942c1bad0fee16f7419df : SUCCESS in 7m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b7c234a628de489c94ea7a2305ffe2a4 : SUCCESS in 1h 15m 12s","accounts_in_message":[],"_revision_number":10},{"id":"6c6b48c31bbb3bfc41e4d87d3d951171d91dfb1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-03 02:09:55.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"123e4afcdb3cd3d4593ebd5ce1069f5c94d407ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 04:19:07.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e3be797967f4d40ada0d2f76a7db4c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/770a04b7d4c947a6871b91b08e91cf01 : SUCCESS in 5m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/870120bd22274207acfec55d966ef02a : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e02a318f9ad488aa7c36d5eb5b2015f : SUCCESS in 8m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77e2c52e0f434fd9b2867f3f3152375b : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3514e93d2b9545f5befafd9bd2fa8958 : SUCCESS in 4m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e8c70ddd14ae46e080f878271af85b51 : SUCCESS in 6m 53s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a6421c31e3a40878fce0eda8b31a542 : FAILURE in 45m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/642c28dfe1884077a43535eca03e9de8 : SUCCESS in 5m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/87086e8746444a92983d6f21848b7761 : FAILURE in 45m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e60eb7f449e94ace8f2332cc301fd58e : FAILURE in 53m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8720716a5eb34706805347e5b905747b : FAILURE in 33m 42s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/6f2017f8125f42698b263dd6b098b2cd : FAILURE in 25m 42s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/450432a71009445c968a1dcb0cd9552e : SUCCESS in 5m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/87482d01b2404e83974aa897a75b09cb : FAILURE in 1h 38m 29s","accounts_in_message":[],"_revision_number":11},{"id":"c8fbfee1503300f3a911d133bbe2933a35a7772f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-09 00:13:29.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c81e7216a14391bb5d30ab7a5627b422579cb0b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 02:06:11.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92e36bb7d83b49d7811b3ce05217ed01\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58061011f52e449fab10f857e22d5191 : SUCCESS in 9m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/477c48eeb8494ad9a5e3c69054e6b30d : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd17c3e3528c4ad0b2cf07f315ebc25f : SUCCESS in 6m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16cb3c2bfadd4b7abafb1be426095e81 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c949f69b7c9e41b8bef9b354bef9261c : SUCCESS in 4m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ffab49ad3cda4d3484168b4545e6d5d8 : SUCCESS in 6m 10s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e77e170a7174a44b0e4f7ad9c121b43 : FAILURE in 1h 14m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/076892b8653d4dcdb8bc2ef27ce5bb7c : SUCCESS in 4m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8242cffe477d4a17b60c7cfc85267a66 : FAILURE in 1h 34m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3b218d0d7b48462388318675c5870fc0 : FAILURE in 1h 50m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1335a8a33ced4b2889c14863430081dd : SUCCESS in 33m 56s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/7780dbe374de42ac96351894f2241eb2 : FAILURE in 25m 08s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d9d2c52072b24ef6b08244565b8dd9c6 : SUCCESS in 4m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bee10586011e43c1bbc15b0c6d7b16b5 : SUCCESS in 54m 50s","accounts_in_message":[],"_revision_number":11},{"id":"ccd1373f31ccbb68cb9a3a6e6f33ddfffb9a6f12","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-09 02:29:39.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"48ac1f81199f1f458740d4bed096f30d6d4a1db9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 04:53:15.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7fab42c11714d4e8370baf13084efc7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4353eec8c36f4861bbd2235dadf1f72f : SUCCESS in 11m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c982d46a40a48c4aa27a15d774e79cd : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc0423ed23e04a0ba98947962768a410 : SUCCESS in 9m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/656ba34fca4048c48f1bdc9284217b03 : SUCCESS in 5m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ede17cb8dae849fc8b8f3240772eeeed : SUCCESS in 5m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7d7cc0c2637345368504a3d680592668 : SUCCESS in 7m 06s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cde953df5dde4bebaece299786143fbd : FAILURE in 1h 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d435c585b5b645f38c95f0e1ecb2c827 : SUCCESS in 6m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/323da5d487bb4f69825fe348bee3b13e : FAILURE in 1h 32m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cf480cf7da7940a9a001a62b26d9b8ba : FAILURE in 2h 10m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7a214eeca00c483d8bca781a04fdce97 : SUCCESS in 34m 00s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/0cbe291ced5f49d7828b738826ec8c08 : FAILURE in 27m 12s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/09316e964ea24d588905c4716461e82a : SUCCESS in 7m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0bb3ed49829e4d44a110e7755fd69a86 : SUCCESS in 1h 42m 33s","accounts_in_message":[],"_revision_number":11},{"id":"38d2d5617e5552f525f512297cde1a66af9bcb59","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-09 20:22:46.000000000","message":"Topic set to neutron-project-id-migration","accounts_in_message":[],"_revision_number":11},{"id":"31c4d486a01aa72c53e01095c213e253cf24076e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-09 22:28:54.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e2e9cd7a89572bc4fe49dc6f855397d18b9d88bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 00:32:32.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c77c5d27de9486f9a5f8ce638b9e017\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c0dfb99f1b44983975166415f32d13c : SUCCESS in 9m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b8d812e7c314fa6baa32885a5345029 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cdaee915d79410bb662d35c923e875c : SUCCESS in 7m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/490ab5a2b5dd4abd9ef4dc98c0d88028 : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9cb4fa3b2664674a89f70e4e159967f : SUCCESS in 5m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a64c3b0ee2174295939c4a32676d87dc : SUCCESS in 9m 28s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9191e2238deb46d692e05cc2ec351a5d : FAILURE in 1h 23m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dba6f579b652435cb9807a41059f2430 : SUCCESS in 4m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0eec9ae416d74cc683c2441390f599ab : SUCCESS in 1h 36m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/01f5743d35af4d8390477a2d744ab61f : SUCCESS in 2h 01m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/578dcd0165274f4dbb40ea44ef823798 : SUCCESS in 34m 38s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/c7930bc6773b46b6b1a32b41ec8ec183 : FAILURE in 19m 15s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/1745e7c87dc247d8916a9147a6ce0f67 : SUCCESS in 8m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8852cd5670b145cc8fb90d15a85fa147 : SUCCESS in 52m 13s","accounts_in_message":[],"_revision_number":11},{"id":"138c0e41c58f724efb1bc2c52c3b6cfe184c0f61","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-10 08:37:22.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":11},{"id":"518eda1c60cce41cc778f565305a7ac1902a53e4","tag":"autogenerated:gerrit:restore","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-10 21:23:58.000000000","message":"Restored","accounts_in_message":[],"_revision_number":11},{"id":"95902eb88183b9ad89f2bab086b5cea0a420148b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 21:25:34.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron-lib for 986960,11","accounts_in_message":[],"_revision_number":11},{"id":"d2317db0ea814f19ff5df0e70421350db1f72509","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-10 21:52:15.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"5cf81470d5ad4c3ce68246f80fe409c48ad2ecda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 23:28:29.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fec8acb86d54be6b8fdd4a01f0d355a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6886bc00fe514193913833be35d6c239 : SUCCESS in 8m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c1222f0cac5484db368334ee42cff27 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90d807f7137c48d193dabc428d744910 : SUCCESS in 8m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/086e38f15dee45b495e33ee6671d5ea7 : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6565afa58ef4fe9ae6066c84e65d2fd : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5f4ca1dbba54422c8f8102d2d6986f57 : SUCCESS in 6m 19s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/4bc3e49f25324a048699b08b1199ea02 : FAILURE in 6m 12s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf13961572ff4fe083e804d4308f56d8 : FAILURE in 57m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e8123d9dddb4381b6943269fec651e8 : SUCCESS in 6m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9ad4e996622e46809199bd3e3e79804a : FAILURE in 21m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a9ca744730704b05a64db0396db0a4a9 : FAILURE in 21m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d69cd109333046b388702e3b2602b286 : FAILURE in 22m 25s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/73b628864dd849b0be612ee9462c1c59 : FAILURE in 27m 06s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3d988ea10f9d4139ba2f98164e186a6f : SUCCESS in 5m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0c371c317863438d987b4f5abe75a0b2 : SUCCESS in 1h 35m 20s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"3e0ed70f70f653cc4003e79ae3a659ba47c37730","revisions":{"73db908f6d5bf513dda2f2fb1bb14861704e16b3":{"kind":"REWORK","_number":1,"created":"2026-05-01 21:43:37.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/60/986960/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/60/986960/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/1"}}},"commit":{"parents":[{"commit":"fe86c1988a2fed008de27b055d0fd127daffaf2a","subject":"Remove logging-resource API definition and documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fe86c1988a2fed008de27b055d0fd127daffaf2a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 14:39:17.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:41:29.000000000","tz":-240},"subject":"WIP: Tenant_id has left the building","message":"WIP: Tenant_id has left the building\n\nRemoved most tenant_id code from neutron-lib, including:\n\n- API attribute code will no longer populate both\n  tenant_id and project_id keys in API requests\n- Any tenant_id attributes in API definitions have been\n  changed to project_id\n- HasProject class can no longer set or get tenant_id\n- get_collection will no longer populate tenant_id\n- resource_fields will no longer populate tenant_id\n- Returned dictionaries will no longer have tenant_id\n\nCleaned-up tests accordingly.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nChange-Id: I74b11c43aedec829e16296a3d074c0b38cd5728d\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/73db908f6d5bf513dda2f2fb1bb14861704e16b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/73db908f6d5bf513dda2f2fb1bb14861704e16b3"}]},"branch":"refs/heads/master"},"523342650f3cb2c1c761008e40a3e920dd180c42":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-05-01 22:20:18.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/60/986960/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/60/986960/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/2"}}},"commit":{"parents":[{"commit":"fe86c1988a2fed008de27b055d0fd127daffaf2a","subject":"Remove logging-resource API definition and documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fe86c1988a2fed008de27b055d0fd127daffaf2a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 14:39:17.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 22:09:53.000000000","tz":-240},"subject":"WIP: Tenant_id has left the building","message":"WIP: Tenant_id has left the building\n\nRemoved most tenant_id code from neutron-lib, including:\n\n- API attribute code will no longer populate both\n  tenant_id and project_id keys in API requests\n- Any tenant_id attributes in API definitions have been\n  changed to project_id\n- HasProject class can no longer set or get tenant_id\n- get_collection will no longer populate tenant_id\n- resource_fields will no longer populate tenant_id\n- Returned dictionaries will no longer have tenant_id\n\nCleaned-up tests accordingly.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/986961\n\nChange-Id: I74b11c43aedec829e16296a3d074c0b38cd5728d\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/523342650f3cb2c1c761008e40a3e920dd180c42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/523342650f3cb2c1c761008e40a3e920dd180c42"}]},"branch":"refs/heads/master"},"d162349e9fb9464ef9786f71e5b9133d3132281b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-05-06 05:03:37.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/60/986960/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/60/986960/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/3"}}},"commit":{"parents":[{"commit":"fe86c1988a2fed008de27b055d0fd127daffaf2a","subject":"Remove logging-resource API definition and documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fe86c1988a2fed008de27b055d0fd127daffaf2a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 14:39:17.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-06 05:02:50.000000000","tz":-240},"subject":"WIP: Tenant_id has left the building","message":"WIP: Tenant_id has left the building\n\nRemoved most tenant_id code from neutron-lib, including:\n\n- API attribute code will no longer populate both\n  tenant_id and project_id keys in API requests\n- Any tenant_id attributes in API definitions have been\n  changed to project_id\n- HasProject class can no longer set or get tenant_id\n- get_collection will no longer populate tenant_id\n- resource_fields will no longer populate tenant_id\n- Returned dictionaries will no longer have tenant_id\n\nCleaned-up tests accordingly.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/986961\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/986967\n\nChange-Id: I74b11c43aedec829e16296a3d074c0b38cd5728d\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d162349e9fb9464ef9786f71e5b9133d3132281b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d162349e9fb9464ef9786f71e5b9133d3132281b"}]},"branch":"refs/heads/master"},"8b7cfa8e102d39a9d937d861ffdf8e6ba52066c0":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-05-06 05:39:35.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/60/986960/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/60/986960/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/4"}}},"commit":{"parents":[{"commit":"fe86c1988a2fed008de27b055d0fd127daffaf2a","subject":"Remove logging-resource API definition and documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fe86c1988a2fed008de27b055d0fd127daffaf2a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 14:39:17.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-06 05:39:02.000000000","tz":-240},"subject":"WIP: Tenant_id has left the building","message":"WIP: Tenant_id has left the building\n\nRemoved most tenant_id code from neutron-lib, including:\n\n- API attribute code will no longer populate both\n  tenant_id and project_id keys in API requests\n- Any tenant_id attributes in API definitions have been\n  changed to project_id\n- HasProject class can no longer set or get tenant_id\n- get_collection will no longer populate tenant_id\n- resource_fields will no longer populate tenant_id\n- Returned dictionaries will no longer have tenant_id\n\nCleaned-up tests accordingly.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/986961\n\nChange-Id: I74b11c43aedec829e16296a3d074c0b38cd5728d\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8b7cfa8e102d39a9d937d861ffdf8e6ba52066c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8b7cfa8e102d39a9d937d861ffdf8e6ba52066c0"}]},"branch":"refs/heads/master"},"c4de9fdd14c6817e385291da06fea5494e04c068":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-05-07 14:20:48.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/60/986960/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/60/986960/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/5"}}},"commit":{"parents":[{"commit":"fe86c1988a2fed008de27b055d0fd127daffaf2a","subject":"Remove logging-resource API definition and documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fe86c1988a2fed008de27b055d0fd127daffaf2a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 14:39:17.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-07 14:20:22.000000000","tz":-240},"subject":"WIP: Tenant_id has left the building","message":"WIP: Tenant_id has left the building\n\nRemoved most tenant_id code from neutron-lib, including:\n\n- API attribute code will no longer populate both\n  tenant_id and project_id keys in API requests\n- Any tenant_id attributes in API definitions have been\n  changed to project_id\n- HasProject class can no longer set or get tenant_id\n- get_collection will no longer populate tenant_id\n- resource_fields will no longer populate tenant_id\n- Returned dictionaries will no longer have tenant_id\n\nCleaned-up tests accordingly.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/986962\n\nChange-Id: I74b11c43aedec829e16296a3d074c0b38cd5728d\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c4de9fdd14c6817e385291da06fea5494e04c068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c4de9fdd14c6817e385291da06fea5494e04c068"}]},"branch":"refs/heads/master"},"69db769724047f1e9e5f98c1c8cd19e994a50c2d":{"kind":"REWORK","_number":6,"created":"2026-05-18 21:57:46.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/60/986960/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/60/986960/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/6"}}},"commit":{"parents":[{"commit":"f7a9e8a542f12ea52929c91b7d69ff23b21f1638","subject":"Document ML2 VLAN tri-state capability behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f7a9e8a542f12ea52929c91b7d69ff23b21f1638"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 14:39:17.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-18 21:57:24.000000000","tz":-240},"subject":"WIP: Tenant_id has left the building","message":"WIP: Tenant_id has left the building\n\nRemoved most tenant_id code from neutron-lib, including:\n\n- API attribute code will no longer populate both\n  tenant_id and project_id keys in API requests\n- Any tenant_id attributes in API definitions have been\n  changed to project_id\n- HasProject class can no longer set or get tenant_id\n- get_collection will no longer populate tenant_id\n- resource_fields will no longer populate tenant_id\n- Returned dictionaries will no longer have tenant_id\n\nCleaned-up tests accordingly.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nChange-Id: I74b11c43aedec829e16296a3d074c0b38cd5728d\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/69db769724047f1e9e5f98c1c8cd19e994a50c2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/69db769724047f1e9e5f98c1c8cd19e994a50c2d"}]},"branch":"refs/heads/master"},"82d9892759545c734c7c148997fc0d0bcff9e0c0":{"kind":"REWORK","_number":7,"created":"2026-06-01 00:18:39.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/60/986960/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/60/986960/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/7"}}},"commit":{"parents":[{"commit":"f7a9e8a542f12ea52929c91b7d69ff23b21f1638","subject":"Document ML2 VLAN tri-state capability behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f7a9e8a542f12ea52929c91b7d69ff23b21f1638"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 14:39:17.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-01 00:08:41.000000000","tz":-240},"subject":"WIP: Tenant_id has left the building, part 1","message":"WIP: Tenant_id has left the building, part 1\n\nRemoved most tenant_id code from neutron-lib, including:\n\n- API attribute code will no longer populate both\n  tenant_id and project_id keys in API requests\n- Added project_id attributes in API definitions, making\n  required_by_policy \u003d True\n- Any tenant_id attributes in API definitions have been\n  changed to required_by_policy \u003d False, will be fully\n  removed in a future change\n- HasProject class can no longer set or get tenant_id\n- get_collection will no longer populate tenant_id\n- resource_fields will no longer populate tenant_id\n- Returned dictionaries will no longer have tenant_id\n\nCleaned-up tests accordingly.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nChange-Id: I74b11c43aedec829e16296a3d074c0b38cd5728d\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/82d9892759545c734c7c148997fc0d0bcff9e0c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/82d9892759545c734c7c148997fc0d0bcff9e0c0"}]},"branch":"refs/heads/master"},"64002fc45ba458f22a611ca6f51d71f13e751d96":{"kind":"REWORK","_number":8,"created":"2026-06-01 03:42:32.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/60/986960/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/60/986960/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/8"}}},"commit":{"parents":[{"commit":"f7a9e8a542f12ea52929c91b7d69ff23b21f1638","subject":"Document ML2 VLAN tri-state capability behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f7a9e8a542f12ea52929c91b7d69ff23b21f1638"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 14:39:17.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-01 03:42:10.000000000","tz":-240},"subject":"WIP: Tenant_id has left the building, part 1","message":"WIP: Tenant_id has left the building, part 1\n\nRemoved most tenant_id code from neutron-lib, including:\n\n- API attribute code will no longer populate both\n  tenant_id and project_id keys in API requests\n- Added project_id attributes in API definitions, making\n  required_by_policy \u003d True\n- Any tenant_id attributes in API definitions have been\n  changed to required_by_policy \u003d False, will be fully\n  removed in a future change\n- HasProject class can no longer set or get tenant_id\n- get_collection will no longer populate tenant_id\n- resource_fields will no longer populate tenant_id\n- Returned dictionaries will no longer have tenant_id\n\nCleaned-up tests accordingly.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nChange-Id: I74b11c43aedec829e16296a3d074c0b38cd5728d\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64002fc45ba458f22a611ca6f51d71f13e751d96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64002fc45ba458f22a611ca6f51d71f13e751d96"}]},"branch":"refs/heads/master"},"036ae747c683b38f2769298802a3e7c08aaeb3dc":{"kind":"REWORK","_number":9,"created":"2026-06-01 17:15:06.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/60/986960/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/60/986960/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/9"}}},"commit":{"parents":[{"commit":"f7a9e8a542f12ea52929c91b7d69ff23b21f1638","subject":"Document ML2 VLAN tri-state capability behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f7a9e8a542f12ea52929c91b7d69ff23b21f1638"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 14:39:17.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-01 17:14:52.000000000","tz":-240},"subject":"WIP: Tenant_id has left the building, part 1","message":"WIP: Tenant_id has left the building, part 1\n\nRemoved most tenant_id code from neutron-lib, including:\n\n- API attribute code will no longer populate both\n  tenant_id and project_id keys in API requests\n- Added project_id attributes in API definitions, making\n  required_by_policy \u003d True\n- Any tenant_id attributes in API definitions have been\n  changed to required_by_policy \u003d False, will be fully\n  removed in a future change\n- HasProject class can no longer set or get tenant_id\n- get_collection will no longer populate tenant_id\n- resource_fields will no longer populate tenant_id\n- Returned dictionaries will no longer have tenant_id\n\nCleaned-up tests accordingly.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nChange-Id: I74b11c43aedec829e16296a3d074c0b38cd5728d\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/036ae747c683b38f2769298802a3e7c08aaeb3dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/036ae747c683b38f2769298802a3e7c08aaeb3dc"}]},"branch":"refs/heads/master"},"58b1c102882cece55ef6e96266dba11b8c3889d1":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2026-06-02 02:08:47.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/60/986960/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/60/986960/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/10"}}},"commit":{"parents":[{"commit":"f7a9e8a542f12ea52929c91b7d69ff23b21f1638","subject":"Document ML2 VLAN tri-state capability behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f7a9e8a542f12ea52929c91b7d69ff23b21f1638"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 14:39:17.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-02 01:34:16.000000000","tz":-240},"subject":"WIP: Tenant_id has left the building, part 1","message":"WIP: Tenant_id has left the building, part 1\n\nRemoved most tenant_id code from neutron-lib, including:\n\n- API attribute code will no longer populate both\n  tenant_id and project_id keys in API requests\n- Added project_id attributes in API definitions, making\n  required_by_policy \u003d True\n- Any tenant_id attributes in API definitions have been\n  changed to required_by_policy \u003d False, will be fully\n  removed in a future change\n- HasProject class can no longer set or get tenant_id\n- get_collection will no longer populate tenant_id\n- resource_fields will no longer populate tenant_id\n- Returned dictionaries will no longer have tenant_id\n\nCleaned-up tests accordingly.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/986962\n\nChange-Id: I74b11c43aedec829e16296a3d074c0b38cd5728d\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/58b1c102882cece55ef6e96266dba11b8c3889d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/58b1c102882cece55ef6e96266dba11b8c3889d1"}]},"branch":"refs/heads/master"},"bbc4abefea1a3d3d257a1fbc4bfc75e05ac2ce5b":{"kind":"REWORK","_number":11,"created":"2026-06-03 02:09:55.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/60/986960/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/60/986960/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/11"}}},"commit":{"parents":[{"commit":"f7a9e8a542f12ea52929c91b7d69ff23b21f1638","subject":"Document ML2 VLAN tri-state capability behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f7a9e8a542f12ea52929c91b7d69ff23b21f1638"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 14:39:17.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-03 02:02:07.000000000","tz":-240},"subject":"WIP: Tenant_id has left the building, part 1","message":"WIP: Tenant_id has left the building, part 1\n\nRemoved most tenant_id code from neutron-lib, including:\n\n- API attribute code will no longer populate both\n  tenant_id and project_id keys in API requests\n- Added project_id attributes in API definitions, making\n  required_by_policy \u003d True\n- Removed tenant_id attributes in API definitions\n- HasProject class can no longer set or get tenant_id\n- get_collection will no longer populate tenant_id\n- resource_fields will no longer populate tenant_id\n- Returned dictionaries will no longer have tenant_id\n\nCleaned-up tests accordingly.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/986962\n\nChange-Id: I74b11c43aedec829e16296a3d074c0b38cd5728d\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bbc4abefea1a3d3d257a1fbc4bfc75e05ac2ce5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bbc4abefea1a3d3d257a1fbc4bfc75e05ac2ce5b"}]},"branch":"refs/heads/master"},"3e0ed70f70f653cc4003e79ae3a659ba47c37730":{"kind":"REWORK","_number":12,"created":"2026-08-10 21:52:15.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/60/986960/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/60/986960/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/60/986960/12"}}},"commit":{"parents":[{"commit":"4a9082c2d3f828bfe5eb03d33bb3edcfcfbf567e","subject":"Merge \"Fix inconsistent hacking version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4a9082c2d3f828bfe5eb03d33bb3edcfcfbf567e"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 14:39:17.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-08-10 21:52:03.000000000","tz":-240},"subject":"WIP: Tenant_id has left the building, part 1","message":"WIP: Tenant_id has left the building, part 1\n\nRemoved most tenant_id code from neutron-lib, including:\n\n- API attribute code will no longer populate both\n  tenant_id and project_id keys in API requests\n- Added project_id attributes in API definitions, making\n  required_by_policy \u003d True\n- Removed tenant_id attributes in API definitions\n- HasProject class can no longer set or get tenant_id\n- get_collection will no longer populate tenant_id\n- resource_fields will no longer populate tenant_id\n- Returned dictionaries will no longer have tenant_id\n\nCleaned-up tests accordingly.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/986962\n\nChange-Id: I74b11c43aedec829e16296a3d074c0b38cd5728d\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3e0ed70f70f653cc4003e79ae3a659ba47c37730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3e0ed70f70f653cc4003e79ae3a659ba47c37730"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
