)]}'
{"id":"openstack%2Fneutron-lib~987372","triplet_id":"openstack%2Fneutron-lib~master~Ia7b0a169c93f87459c996b32027bb78a68ae34ff","project":"openstack/neutron-lib","branch":"master","topic":"bug/2150864","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-05-06 17:21:59.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-05-06 20:27:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia7b0a169c93f87459c996b32027bb78a68ae34ff","subject":"sfc: Add is_filter and is_sort_key to SFC API","status":"MERGED","created":"2026-05-05 14:54:44.000000000","updated":"2026-05-06 20:28:44.000000000","submitted":"2026-05-06 20:27:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987372-bug/2150864","meta_rev_id":"13dbac2c5fe6321dedcd54487de995e375de869e","_number":987372,"virtual_id_number":987372,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-06 20:27:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-05-06 17:21:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-05-06 18:17:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-05-06 18:17:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-05-06 17:21:59.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-05 16:40:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-05 18:32:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-05-05 21:09:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-05-06 11:12:25.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"d129cda0fdf7089c576f4b37d6fbd7f68678b3da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-05 14:54:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"562b7bd696cce369712c61b1e4b7fe93f58d5d17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 16:40:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a05ec0666b4b479db195479daf48b267\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/525b0afad0364061abd7a153935e3991 : SUCCESS in 9m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6df2b7a452a9494eaec7855710f5ca75 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bfa2aad621b455a8ffc775026122423 : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aef16d3b12944bbfac0ba20e33dadc82 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb8fa0af11e647e69eacd3be822c2735 : SUCCESS in 5m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b99de8fde0a348609dccbdc84f0f4238 : SUCCESS in 5m 55s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fd3fc179a504d949a6205c812275d57 : SUCCESS in 1h 16m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8fdece6a1fc346d897a245124db67c82 : SUCCESS in 1h 23m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a15fc18ececf4229b6f87cb8ba990856 : SUCCESS in 1h 42m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d20ab16287ae492fb90cbb663fc70eb2 : SUCCESS in 31m 46s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/20576703171e4f15b4b5337b1b4ff3f4 : SUCCESS in 21m 34s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0ffc021c1eb842f9987ce4238a37c90b : SUCCESS in 4m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6205275c1e6f4a0fbc7097ee7e2a16d2 : SUCCESS in 52m 52s","accounts_in_message":[],"_revision_number":1},{"id":"cd832078fbeea7778cc1c0af8bfdb0b962bfba0d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-05 18:32:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e56178f3731450548b0d1f74706eb32fbea12f7a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-05 21:09:34.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84070caaa70ef3d1b31485765c05d46895e72a58","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-06 11:12:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6db960840dd32cf32e6f18180bd8d5d51e425f1e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-06 11:12:43.000000000","message":"Patch Set 1: -Workflow Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eed0ed95301c4876870332c911e7773eb3141962","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 11:12:44.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":1},{"id":"88a1a455a2f7bb63639ae5a0d745715e5c0177fb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-06 12:54:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd3bc6891168befcd6df94767ecef704bd3a3132","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-06 13:12:07.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"d97d2b1995eeb28e2d25885b13386f2c98198563","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-06 14:20:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8dbdd8aa7e4d46cc17ac8b1a39da8311688da5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-06 15:15:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"aecf63e9b04fe345096b059227fe08de1a5063c5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-06 15:16:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a8af447745d15c4e55e1571b2822eb4bb8065e9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 17:14:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3debdfba6df94ce0b8f319b29abeeee2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aee96e9410b64c35ac39b1ab543b825d : SUCCESS in 8m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f41c1fdbb4b44654846080b6d8e87ea4 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/802b9c996ce348b4a20b92caa3dd0317 : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b1ae87ef570e4e05869dd3f96b549f59 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/906db72cb8cb498eaccaebadec8bd0a8 : SUCCESS in 3m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/27b328c2de4b45e1a4cbb79bbc52b0eb : SUCCESS in 6m 37s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/334862ca60f94f63a227515c067ca9d3 : SUCCESS in 1h 51m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96f30fb9c9d64da18024c60c98d6d5e6 : SUCCESS in 3m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0ab06f06f1dc46848dd01dbffba922a2 : SUCCESS in 1h 19m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/52947f149c7f4b75800c97e09179d6d0 : SUCCESS in 1h 57m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f65e1bcc48184d90807eddb97ecad0e9 : SUCCESS in 34m 10s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/09178b7093654486829eb930e3beef73 : SUCCESS in 22m 03s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9ed7d6ace9f445e291cc932512be8547 : SUCCESS in 6m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/22f83e02b59f40dbb7963abfed24cb2c : SUCCESS in 1h 31m 38s","accounts_in_message":[],"_revision_number":2},{"id":"88d42bc136e44c70e7349be98a5d2ff2b587cca7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-06 17:21:59.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"58f31812527a69ab4868794cfb4b53e026139b5f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-06 18:17:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d2ebf59a3d4fa2f09d91b7e8386f93a8b0318b6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 18:17:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"41c656c0c239c91b0e7f316a847747ed08d20d9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 20:27:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec37f7e3b9a54e17bcf89bc7dfdd583f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8962c558a0f848bb9212fb3b070d0c10 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51c191bd69a341c3b421592cc91fc640 : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be6b3cd50968456889ab6febec30e367 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d58dd338fa742abaab610a003722a93 : SUCCESS in 2m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f946f7eefbd346ba8f1a4d0ea49d07bb : SUCCESS in 1h 26m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bd0aefae3704a3593634761487668b5 : SUCCESS in 3m 26s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6d2f62e22812477b816585e7ad05867e : SUCCESS in 1h 22m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b75eb77753ea475abf43e0f06d2558b4 : SUCCESS in 2h 09m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a7f93f0cdfcb4216a6a0e696368b4d58 : SUCCESS in 1h 25m 42s","accounts_in_message":[],"_revision_number":2},{"id":"3ce2810186061885bfefe115a7ee197b8b645d35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 20:27:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"13dbac2c5fe6321dedcd54487de995e375de869e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 20:28:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c28ea87152a45c4ad58b2d53a8ed30f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7af8a0dc02b47aa938b57207e91c3a8 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/408ca1dc89664e0493ae29f87a70c7f6 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d9852d36cabae82a52cbc9b2d4f499ec9cc1e784","revisions":{"3f9942c5aac9c8626cfad76105a3581a74a826a5":{"kind":"REWORK","_number":1,"created":"2026-05-05 14:54:44.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/987372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/72/987372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/72/987372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/72/987372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/72/987372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/72/987372/1"}}},"commit":{"parents":[{"commit":"ea424098bb1f44bd28732c3f8285022eaa6a15d7","subject":"Merge \"evpn: Add EVPN API definitions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ea424098bb1f44bd28732c3f8285022eaa6a15d7"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-05-04 11:21:04.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-05-05 14:44:39.000000000","tz":120},"subject":"sfc: make name fields filter in API","message":"sfc: make name fields filter in API\n\nThe patches pushed for #2142034 (see [1]) made filter\nvalidation enabled for all extensions. The networking-sfc\nextensions had no is_filter attribute, this patch\nadds that to the name field as that is tested in\nunit tests, so I assume was expected to work at least.\n\nThe sfc API deifinitions\u0027 migration was never finished\nso [1] is fixing the gate by adding the same filter for\nAPI extensions in networking-sfc repository.\n[2] was the original patch to finish the migration of the\nextensions from sfc repository to neutron-lib.\n\n[1]: https://review.opendev.org/c/openstack/networking-sfc/+/987371\n[2]: https://review.opendev.org/c/openstack/networking-sfc/+/553819\n\nRelated-Bug: #2142034\nRelated-Bug: #2150864\nChange-Id: Ia7b0a169c93f87459c996b32027bb78a68ae34ff\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3f9942c5aac9c8626cfad76105a3581a74a826a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3f9942c5aac9c8626cfad76105a3581a74a826a5"}]},"branch":"refs/heads/master"},"d9852d36cabae82a52cbc9b2d4f499ec9cc1e784":{"kind":"REWORK","_number":2,"created":"2026-05-06 15:15:08.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/987372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/72/987372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/72/987372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/72/987372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/72/987372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/72/987372/2"}}},"commit":{"parents":[{"commit":"ea424098bb1f44bd28732c3f8285022eaa6a15d7","subject":"Merge \"evpn: Add EVPN API definitions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ea424098bb1f44bd28732c3f8285022eaa6a15d7"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-05-04 11:21:04.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-05-06 15:14:05.000000000","tz":120},"subject":"sfc: Add is_filter and is_sort_key to SFC API","message":"sfc: Add is_filter and is_sort_key to SFC API\n\nThe patches pushed for #2142034 (see [1]) made filter\nvalidation enabled for all extensions. The networking-sfc\nextensions had no is_filter attribute, this patch\nadds that to the name field as that is tested in\nunit tests, so I assume was expected to work at least.\n\nThe sfc API deifinitions\u0027 migration was never finished\nso [1] is fixing the gate by adding the same filter for\nAPI extensions in networking-sfc repository.\n[2] was the original patch to finish the migration of the\nextensions from sfc repository to neutron-lib.\n\n[1]: https://review.opendev.org/c/openstack/networking-sfc/+/987371\n[2]: https://review.opendev.org/c/openstack/networking-sfc/+/553819\n\nRelated-Bug: #2142034\nRelated-Bug: #2150864\nChange-Id: Ia7b0a169c93f87459c996b32027bb78a68ae34ff\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d9852d36cabae82a52cbc9b2d4f499ec9cc1e784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d9852d36cabae82a52cbc9b2d4f499ec9cc1e784"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
