)]}'
{"id":"openstack%2Fneutron-lib~987378","triplet_id":"openstack%2Fneutron-lib~master~I8f566a5aeaf9daf29dce8acd3db76c34d84ba7eb","project":"openstack/neutron-lib","branch":"master","attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-05-06 13:08:21.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-05-06 13:08:21.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I8f566a5aeaf9daf29dce8acd3db76c34d84ba7eb","subject":"Update definitions to support name field filter","status":"NEW","created":"2026-05-05 15:51:21.000000000","updated":"2026-05-06 13:08:43.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":14,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"7480e4f5c419fae9999cbbc3d80e31d1210d89a9","_number":987378,"virtual_id_number":987378,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-05 20:32:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"all":[{"value":1,"date":"2026-05-06 11:34:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-05 16:46:06.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2026-05-05 18:03:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-05 18:09:14.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-05-06 11:34:54.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2026-05-06 13:08:43.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"}],"messages":[{"id":"1ecb7276fc1291dbc097e41287dfe807100e9724","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-05 15:51:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e0b2dbcb437c031d57f41a4b70b00dc86056e76","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-05 16:46:06.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6005918783d560b5d2c2934808f8414d3e973bad","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-05 16:46:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5bddbcb3d29d6ce328ae3dccf778771fba82d459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 18:03:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8459a42bea744ea990c3c0b83bded8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a03308b6a3c24d388bbd861a90c8a0ea : SUCCESS in 4m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef74765dbd714cd19e020a5e9a1afcce : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b592573b3b7544238b88a3390e1131a1 : FAILURE in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e75c2c7a586479894b1d07fda159ccd : SUCCESS in 3m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3ff7ed07b8cd49acb56ec07f8ecbc44b : SUCCESS in 3m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/90f3e6bb9fd44a4b8524926e91b34b1a : SUCCESS in 6m 38s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c77b3f8dda63472980ace23ee2bc44f9 : SUCCESS in 1h 31m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6a2729c2489945ad84513cf7bbdede6d : SUCCESS in 1h 29m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fb9530bd799f4135aa9eed138fcc2611 : SUCCESS in 2h 09m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2e478aaf64a34556bdceb1eb478b7bdb : SUCCESS in 34m 44s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/329601b01fbb4195b0996bb41efa9e26 : SUCCESS in 27m 25s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6e8aa2c8f29749cea50bcae058e8f4a2 : SUCCESS in 5m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4771993d9a484bb8b42be4ea2c000108 : SUCCESS in 1h 21m 40s","accounts_in_message":[],"_revision_number":1},{"id":"9326c2c7fd19d35c5445e2c2a928f568929e7b4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-05 18:15:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2c61938d7c7664219b0be654119ea371e4179884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 20:32:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14ef2b0974644acd843cccc38d169f7e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0aa4c920052412cbe8ace55f30c8c4f : SUCCESS in 7m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fb833ae313d4a09b5fb39c9fc735b6e : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81e678e83f2541b0b0ce5e487d08481f : SUCCESS in 5m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/365091ac0b4f47fda81ecbd58d759f4d : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/847bfda176a64398b9e49826c42c7d2d : SUCCESS in 3m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/da7be89ea417446388429bb89a30f778 : SUCCESS in 4m 14s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5394b7157ea44f9ad19c1d981e6fd0b : SUCCESS in 1h 59m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/044f818f35a749ea99beae2f77b25e3d : SUCCESS in 1h 19m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2ec0097155074748818b52de61b54f5a : SUCCESS in 2h 15m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9a03492775744a049e2501148bb3d364 : SUCCESS in 33m 29s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/f336bbfd8b6544619307fd121313ad5d : SUCCESS in 29m 16s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/df86cd1e79034f2b9ebed253864369ac : SUCCESS in 3m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b6ef6f27c696441c96cf3ec504d17e8e : SUCCESS in 43m 16s","accounts_in_message":[],"_revision_number":2},{"id":"38c02232e2fcda650c1d5e617fbb8fc8bd153812","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-05 21:10:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a73b29bbb2fd586f8d46c29c6b343294e6c67d7b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-05 21:11:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d933c112400a4e4c72d9ff32e76443c4f675ca6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 23:20:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd69d8d86b974981aee6e129a1c9052b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb3a6800b80a4f45bd4eb503a86a260e : SUCCESS in 8m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c111c58e1d74928afb12b9dfe7f2891 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a82f05a9f684d21b6c2ddef727ffe70 : SUCCESS in 6m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5878580b73c44336bd7623561d626285 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f59c388f52874c0d9a3aef4fbca3fe99 : SUCCESS in 3m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1a3a496d53a84aa6ae5373baae1266e2 : SUCCESS in 6m 31s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2613ada9e1c94fa18d999966dbd8cadf : SUCCESS in 2h 05m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b9645a1edce4435abf370bb2eca3463d : SUCCESS in 1h 28m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/344fb99e03ab41a396118cb160474ce2 : SUCCESS in 2h 05m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f7b49326d8104b02bc2e1db96f9ebeae : SUCCESS in 33m 33s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/7fb44f4d404e4f489cf4855f50f0facb : SUCCESS in 16m 59s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/85bd09e079114d44ba49954feb655a5e : SUCCESS in 7m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/408ef6536931438b988b6d0849c0c351 : SUCCESS in 1h 07m 48s","accounts_in_message":[],"_revision_number":2},{"id":"ce99170e7ce8549d37ae049624e70ed86bffc48b","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-05-06 11:34:54.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"41628362e94be9654ea5d427bcf67fe6bdda6146","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-06 12:02:55.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"25500af2ec1ead6860c7393a4cc9cd315e7d8265","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-06 13:08:21.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b888b84ee21ceb2e4a147ec2e5283553e7ff52b3","revisions":{"b67dd10a6fb081a7193a8e1eaf9fb334258e33b5":{"kind":"REWORK","_number":1,"created":"2026-05-05 15:51:21.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/78/987378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/78/987378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/987378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/987378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/987378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/78/987378/1"}}},"commit":{"parents":[{"commit":"c5980cda3ad179ceb744ea1ebda3bc41ec04586e","subject":"Merge \"Remove fip64 API definition and documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c5980cda3ad179ceb744ea1ebda3bc41ec04586e"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-05 15:48:14.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-05 15:48:14.000000000","tz":-240},"subject":"Update definitions to support name field filter","message":"Update definitions to support name field filter\n\nAs was done with networking-sfc, update the name fields\nin firewall_v2, taas, tap_mirror and vpn definitions\nto support filtering based on \u0027name\u0027.\n\nRelated-bug: #2142034\n\nChange-Id: I8f566a5aeaf9daf29dce8acd3db76c34d84ba7eb\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b67dd10a6fb081a7193a8e1eaf9fb334258e33b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b67dd10a6fb081a7193a8e1eaf9fb334258e33b5"}]},"branch":"refs/heads/master"},"b888b84ee21ceb2e4a147ec2e5283553e7ff52b3":{"kind":"REWORK","_number":2,"created":"2026-05-05 18:15:44.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/78/987378/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/78/987378/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/987378/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/987378/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/987378/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/78/987378/2"}}},"commit":{"parents":[{"commit":"c5980cda3ad179ceb744ea1ebda3bc41ec04586e","subject":"Merge \"Remove fip64 API definition and documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c5980cda3ad179ceb744ea1ebda3bc41ec04586e"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-05 15:48:14.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-05 16:45:30.000000000","tz":-240},"subject":"Update definitions to support name field filter","message":"Update definitions to support name field filter\n\nAs was done with networking-sfc, update the name fields\nin firewall_v2, taas, tap_mirror and vpn definitions\nto support filtering based on \u0027name\u0027.\n\nRelated-bug: #2142034\n\nChange-Id: I8f566a5aeaf9daf29dce8acd3db76c34d84ba7eb\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b888b84ee21ceb2e4a147ec2e5283553e7ff52b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b888b84ee21ceb2e4a147ec2e5283553e7ff52b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
