)]}'
{"id":"openstack%2Fneutron-lib~987437","triplet_id":"openstack%2Fneutron-lib~master~I2364add1937e4947bd3b8384b066bd5f4fd25737","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-05-06 13:25:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2364add1937e4947bd3b8384b066bd5f4fd25737","subject":"remove reference to non-existent method","status":"MERGED","created":"2026-05-05 22:12:30.000000000","updated":"2026-05-06 13:26:12.000000000","submitted":"2026-05-06 13:25:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987437","meta_rev_id":"a1778315da6eb049cff1dd59a0433b3a429b1692","_number":987437,"virtual_id_number":987437,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-06 13:25:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-05-06 11:31:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-05-05 22:26:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2026-05-06 11:31:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-05 22:26:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-05-06 00:03:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-06 11:31:41.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"3ba34e05fe8a9b86f8b80a2a3f51dfbd21f0c28a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-05-05 22:12:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8dd4c7a4332f2051a58755d8d94ebcaa766646d9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-05 22:26:23.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bca8f3091f534b63afb2f9b414ea831987992988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 00:03:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cb084b1d4514ed291f0aaf42d15ad66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/880d5a00e950453ba0528d8d8675d226 : SUCCESS in 8m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d49f0ab5f47421d888e83c997d09aeb : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b137072ce8174ac1aacbc523517b2760 : SUCCESS in 6m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f2717fb66b94c1ab75290c87e5b44ca : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa6c7f8834374b259c8d1212b9330c45 : SUCCESS in 2m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/213a90586c2249a9be1137f7d55f290d : SUCCESS in 5m 40s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6679f7e7de0f4f0abc4fb4bb5b00fbc7 : SUCCESS in 1h 07m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/68929c459b7a44a9b30c6dd1bc1b0043 : SUCCESS in 1h 24m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d2f738baaef249fb8d17e79d385a7920 : SUCCESS in 1h 50m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4edf80a97d334c138a8d3c78d6e1f476 : SUCCESS in 31m 59s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/7ad29055861e419da22b99e9c759b134 : SUCCESS in 26m 52s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/4603750b081a42148fdc07809b958559 : SUCCESS in 5m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cd7a762301574afdb08331e4a188c72f : SUCCESS in 1h 31m 12s","accounts_in_message":[],"_revision_number":1},{"id":"73b3b7f53082f2fbf709adcd6853f32cab421a76","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-06 11:31:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c648b1811c8f88cecd4b650f23b2b056f2c847c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 11:32:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81a3ce63ebd80b4bafc4aa73072c8eb6cbfcdd12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 13:25:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8386a7c7ac094c918f11afec6a41ee2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d1616873e724cdfbdc07fad81cc49e2 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b9f9ce7cf24406c90a557f024ce3cf2 : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0f4728df7082489db5b382757e6cc65e : SUCCESS in 2m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c006305decea4ed8b9e074bf718d988e : SUCCESS in 3m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0bc1409aac97496894f26ede4e38ca5e : SUCCESS in 1h 09m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e4f47c00feb44c10898f108ba7edf203 : SUCCESS in 1h 20m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/336e3af7cfa342b2be89c831802f4b38 : SUCCESS in 1h 52m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/370cb08a2ff54d77ac54cf1d7b39d217 : SUCCESS in 43m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d6c1d3aa5cfe89ba72c2b499f2e9ca01f4f13f5a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 13:25:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a1778315da6eb049cff1dd59a0433b3a429b1692","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 13:26:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6496ab4910e475f8d809ffbc0189d60\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c3f49aac4e74055b05ff857188d9c19 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83b270868cb96cf689ddd0f8a230d7c3e2767e76","revisions":{"83b270868cb96cf689ddd0f8a230d7c3e2767e76":{"kind":"REWORK","_number":1,"created":"2026-05-05 22:12:30.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/37/987437/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/37/987437/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/37/987437/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/37/987437/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/37/987437/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/37/987437/1"}}},"commit":{"parents":[{"commit":"c5980cda3ad179ceb744ea1ebda3bc41ec04586e","subject":"Merge \"Remove fip64 API definition and documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c5980cda3ad179ceb744ea1ebda3bc41ec04586e"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-05-05 22:10:27.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-05-05 22:12:28.000000000","tz":-300},"subject":"remove reference to non-existent method","message":"remove reference to non-existent method\n\nThe get_allocation() method does not exist on any Neutron TypeDriver\u0027s\nand exists on on those inheriting from the tunnel abtract ML2TypeDriver\nbut is never called so remove it from being mentioned.\n\nChange-Id: I2364add1937e4947bd3b8384b066bd5f4fd25737\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/83b270868cb96cf689ddd0f8a230d7c3e2767e76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/83b270868cb96cf689ddd0f8a230d7c3e2767e76"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
