)]}'
{"id":"openstack%2Fneutron-lib~987769","triplet_id":"openstack%2Fneutron-lib~master~I4e4baf909a4a43b3e0d21662195b79e911d0e242","project":"openstack/neutron-lib","branch":"master","attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-07-07 20:13:14.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-07-07 04:33:39.000000000","reason":"Vote got outdated and was removed: Code-Review+1"}},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-08-04 10:16:24.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I4e4baf909a4a43b3e0d21662195b79e911d0e242","subject":"add initial typing to TypeDriver interface","status":"NEW","created":"2026-05-07 22:13:20.000000000","updated":"2026-08-04 10:16:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":16,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"b97912bc202a38f8897a2bc9b811dd23e41cf98c","_number":987769,"virtual_id_number":987769,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-07 07:14:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-08 00:05:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-11 09:43:45.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-05-26 23:09:07.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-05-26 23:09:07.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-06-02 16:44:39.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"cfd0a4d02319249b262336ac2294f688318e2305","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-05-07 22:13:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a47329040565025d56b701508dc449f5c454e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 00:05:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/633bc88c4c4e477e97a2aae61a547318\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb80cafbee3247d697633736fb7f3c27 : SUCCESS in 8m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd59e63943de4c06a85901c4420bb620 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae0466f0e83e4e0dafbe898c24c27562 : SUCCESS in 6m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9aba71c5603f44d5b65b0686ee30c760 : SUCCESS in 2m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1bd58d3194341d5972996a89c21050c : SUCCESS in 3m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cc08bf1ccc1c425f97c9f7badd7a9c97 : SUCCESS in 4m 07s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b2ef46338374a9ea7034138ae3c563c : SUCCESS in 1h 01m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/078653767eaf4a789a2cba0dfe049013 : SUCCESS in 1h 24m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ae900c123a704711ba2c9c219b5770db : SUCCESS in 1h 51m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1ea6dee4423141718fd0f8e57e9cddd4 : SUCCESS in 33m 11s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/10f121c095cd4daca54e8e1218e6b2ac : SUCCESS in 20m 56s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/64597b9cb53343d8b1980823041a5e83 : SUCCESS in 5m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c920b38a8d014795abad4b94c1d1e2e2 : SUCCESS in 1h 26m 27s","accounts_in_message":[],"_revision_number":1},{"id":"08eb6b9b46325d1a8a12ee9fc7bfdcf7e8fee834","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-11 09:43:45.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f8c037d543498d9b319a76813125149680d1e99","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-05-11 12:07:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"706bc5c63a698f4dda76938203ee8e442c4b2c7f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-02 16:44:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8980817c97b0092add65717e337d93dea5efe3a3","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-06-02 23:22:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6baf63ffd3ea7b4aa0b1ac7d4334df117df142d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-07-07 04:33:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7455f053151f5aa62c2653b294ed5726ef82dfe9","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-07-07 04:33:56.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3c8ea50db9ff99bbac8aebe9f5f04efa20cf61e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-07-07 05:07:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ce73dc707f1d183c06f8a2c2def6473ef6785075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 07:14:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aebd918bf04a4cd68e60c47a24cc8947\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69ad7e8792614e95b5fc02780b1e6768 : SUCCESS in 8m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d9c2b713caa4a0bb31162950b66e0f5 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6e2b8982400451fbef85b01ffde7f38 : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3738b1b3f184ab8a8ebed28a6c9421d : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8e703dbe9534b35bffdcb5d748204f8 : SUCCESS in 2m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0515a4dcd7774d59a061a6805d5e5674 : SUCCESS in 3m 25s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/fcfa4637ea0f4aa0b8ca003574d40d76 : FAILURE in 3m 09s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/119c0539230c4e88884beb97a32ece27 : SUCCESS in 1h 22m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5846f460bc0d4870a74d89ec66409887 : SUCCESS in 4m 58s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dc3359712e354f899fcc7e2b03236296 : SUCCESS in 1h 28m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9cd789e423a24a7da50a04b4067554f7 : SUCCESS in 2h 05m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6c17ac0f315e4a51bb9aa2a0c2e9bd42 : SUCCESS in 31m 15s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/9282396f926f4b3fa3ef160347899831 : SUCCESS in 18m 30s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d25a608164c443059c175a18a3327eef : SUCCESS in 3m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f54cb0e44ba24b3888fdbb1d7e25d1aa : SUCCESS in 46m 35s","accounts_in_message":[],"_revision_number":3},{"id":"edf8e5483c5854396c993ab62094c00e3b55d847","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-07 20:13:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b97912bc202a38f8897a2bc9b811dd23e41cf98c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-04 10:16:24.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9c0728be803a11c18367624b75350b35c7bae23d","revisions":{"f025802592484a27cfff6746fc0c82faeaf89792":{"kind":"REWORK","_number":1,"created":"2026-05-07 22:13:20.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/69/987769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/69/987769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/987769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/987769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/987769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/69/987769/1"}}},"commit":{"parents":[{"commit":"e3dacf2eb28760a28138014b7202ed2f816b2cd5","subject":"Merge \"Remove router-interface-fip API definition and documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e3dacf2eb28760a28138014b7202ed2f816b2cd5"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-05-07 22:07:34.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-05-07 22:12:52.000000000","tz":-300},"subject":"add initial typing to TypeDriver interface","message":"add initial typing to TypeDriver interface\n\nAdded some initial typing to the TypeDriver interface to assist with\nvalidations of the data supplied.\n\nChange-Id: I4e4baf909a4a43b3e0d21662195b79e911d0e242\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f025802592484a27cfff6746fc0c82faeaf89792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f025802592484a27cfff6746fc0c82faeaf89792"}]},"branch":"refs/heads/master"},"f15ce817f1458aa4072e1f1c44f19e5f14ef78e2":{"kind":"REWORK","_number":2,"created":"2026-07-07 04:33:39.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/69/987769/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/69/987769/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/987769/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/987769/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/987769/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/69/987769/2"}}},"commit":{"parents":[{"commit":"2df8ca85c993ca20cc66c50882c3d69f5ee55ed2","subject":"Merge \"Add is_filter to PVLAN attributes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2df8ca85c993ca20cc66c50882c3d69f5ee55ed2"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-05-07 22:07:34.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-07-07 04:30:04.000000000","tz":-300},"subject":"add initial typing to TypeDriver interface","message":"add initial typing to TypeDriver interface\n\nAdded some initial typing to the TypeDriver interface to assist with\nvalidations of the data supplied.\n\nChange-Id: I4e4baf909a4a43b3e0d21662195b79e911d0e242\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f15ce817f1458aa4072e1f1c44f19e5f14ef78e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f15ce817f1458aa4072e1f1c44f19e5f14ef78e2"}]},"branch":"refs/heads/master"},"9c0728be803a11c18367624b75350b35c7bae23d":{"kind":"REWORK","_number":3,"created":"2026-07-07 05:07:37.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/69/987769/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/69/987769/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/987769/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/987769/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/987769/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/69/987769/3"}}},"commit":{"parents":[{"commit":"2df8ca85c993ca20cc66c50882c3d69f5ee55ed2","subject":"Merge \"Add is_filter to PVLAN attributes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2df8ca85c993ca20cc66c50882c3d69f5ee55ed2"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-05-07 22:07:34.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-07-07 04:56:20.000000000","tz":-300},"subject":"add initial typing to TypeDriver interface","message":"add initial typing to TypeDriver interface\n\nAdded some initial typing to the TypeDriver interface to assist with\nvalidations of the data supplied.\n\nChange-Id: I4e4baf909a4a43b3e0d21662195b79e911d0e242\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9c0728be803a11c18367624b75350b35c7bae23d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9c0728be803a11c18367624b75350b35c7bae23d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
