)]}'
{"id":"openstack%2Fneutron-lib~988601","triplet_id":"openstack%2Fneutron-lib~master~I83c5f199fb0eae3317457057f4f314f96a67e01d","project":"openstack/neutron-lib","branch":"master","topic":"qinq_0x88a8","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-05-15 16:18:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I83c5f199fb0eae3317457057f4f314f96a67e01d","subject":"Fix QinQ tag 0x88a8 in documentation","status":"MERGED","created":"2026-05-14 11:14:52.000000000","updated":"2026-05-15 16:19:57.000000000","submitted":"2026-05-15 16:18:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988601-qinq_0x88a8","meta_rev_id":"ac1945b562e032cab13333bc06819b49e2c6cf21","_number":988601,"virtual_id_number":988601,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-15 16:18:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2026-05-15 14:08:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-05-14 13:10:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2026-05-15 14:08:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-14 13:10:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-05-14 13:26:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-14 13:49:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-05-14 13:49:23.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-05-15 14:08:04.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"b7b5246bdbcadad9225dede1c00974163179addf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-14 11:14:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35a221bce5baafe517e2648d572999b726814c4d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-14 13:10:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"66bc84c180cf6c40b2b215b25da6fc8321320f0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 13:26:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1de6487267c24367be94eb86ed64759f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/42502196a170451fa118ff85e889c8f6 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b48c5b1c86944eaa9f8aee8eb99492d2 : SUCCESS in 10m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9468fbb8ad8c465a8364d5801cd6c248 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99e32b76f02a4a35a81bd138e3940ff5 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/117bb4e1313c4fb1909b666955747107 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d6904d693bf04334ad97dd1d6250771f : SUCCESS in 2m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/35e64d6b999d4e2ab1d4cb7955a66508 : SUCCESS in 3m 33s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43485fb5c74f4fb0b80847f4bc83de23 : SUCCESS in 1h 47m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/815ec6b963264752babb90a632c108a4 : SUCCESS in 4m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/37e45621a02140f189dab6fc06b3e169 : SUCCESS in 1h 43m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5d735b4a2422432aa96a938c0aee5fed : SUCCESS in 2h 06m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8af9368949464a049d4ffb6cd6651e74 : SUCCESS in 34m 36s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/8ea25bfa3a7c4f2f885ffc974621bb47 : SUCCESS in 21m 59s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/bd28989910ba44fbb8186986bbd08f62 : SUCCESS in 4m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d00dd05329014886bf0f71f4e3f8dac2 : SUCCESS in 1h 30m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8560f6525d596a9d78145f5698b391dec165d611","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-05-15 14:08:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80b7eaf1a43cc678fdf1b732b4a97f0c2b78b065","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 14:08:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf59597e4dfcdf81cda9c598e468fbfc5c931c1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 16:18:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/970837720f504079949268b205c23d9d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dc90cb28c9ba401a9192122c8b45b7b9 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6574afa27e94adca5a5142dd7c88795 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/312d94933855492f933bbd57e24951b8 : SUCCESS in 8m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0f4246ab248a4ea787b7cadb0a521212 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a619c1f20ef4cae8f6150aa1d9c5e5c : SUCCESS in 3m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bbaba56559f346f1bc9867986fa23fdb : SUCCESS in 1h 51m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6212256e4d14047af2bf9fc239bd38b : SUCCESS in 6m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8818bd61c60e40bab47b84e337523d53 : SUCCESS in 1h 29m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f12c012766e34b64b2a21d74389c5843 : SUCCESS in 2h 09m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a708fcf24d174149b1124950f8683dc2 : SUCCESS in 56m 29s","accounts_in_message":[],"_revision_number":1},{"id":"291c63c75e46bbc4f69ab5fd442938b21846c709","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 16:18:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ac1945b562e032cab13333bc06819b49e2c6cf21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 16:19:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a207a5bfa64645fab96785ac39290b94\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0d8454f301ce42ee9acf566e0fef5a36 : SUCCESS in 57s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c9f22c00a0d41ba9ccc31910d883a2a : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ff1ea4024954125b2c421a81afa4fa4 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18b45201ffc3e3a796d9853d6121ab35732c3171","revisions":{"18b45201ffc3e3a796d9853d6121ab35732c3171":{"kind":"REWORK","_number":1,"created":"2026-05-14 11:14:52.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/988601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/01/988601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/01/988601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/01/988601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/01/988601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/01/988601/1"}}},"commit":{"parents":[{"commit":"083afc26e27744bbeda3e415149f017e0b39190c","subject":"Merge \"Add security-groups-default-statefulness API definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/083afc26e27744bbeda3e415149f017e0b39190c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-14 11:13:03.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-14 11:14:27.000000000","tz":120},"subject":"Fix QinQ tag 0x88a8 in documentation","message":"Fix QinQ tag 0x88a8 in documentation\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I83c5f199fb0eae3317457057f4f314f96a67e01d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/18b45201ffc3e3a796d9853d6121ab35732c3171"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/18b45201ffc3e3a796d9853d6121ab35732c3171"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
