)]}'
{"id":"openstack%2Fneutron-lib~989019","triplet_id":"openstack%2Fneutron-lib~master~I044a1a3aaa92f9d05e68bd9a1d3d67a9ce3e2798","project":"openstack/neutron-lib","branch":"master","topic":"pbr-runtime-rm","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-07-07 04:37:38.000000000","reason":"Change was abandoned"},"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-07-07 04:37:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I044a1a3aaa92f9d05e68bd9a1d3d67a9ce3e2798","subject":"remove pbr as runtime dependency","status":"ABANDONED","created":"2026-05-18 18:54:01.000000000","updated":"2026-07-07 04:37:38.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dc6e81f8bf2fc67704d613e8767694413424869c","_number":989019,"virtual_id_number":989019,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-05-20 00:45:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-18 20:49:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-19 08:09:11.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-05-19 17:38:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"}],"messages":[{"id":"1bc1db4ca6425100a509c55699f1f84d93c3e10a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-05-18 18:54:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04ca8c05fffdd5354d18425daae67bcd83add26f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 20:49:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/befebd10b6e240dcbfbfc4c2a3a2b0b3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/31469e27ad224e49881ac2d19b27717e : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7c2b5ab5ccd4939bc933f143fe72fd2 : SUCCESS in 10m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ada2b6c503f64fd3959be9609a0cc977 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a5b70849f234b2cbc8c7f19c83875fc : SUCCESS in 8m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5447212f41634110b7bc90636fdb1488 : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9843d6b2c1864eacbf1b17c599e14e68 : SUCCESS in 5m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d9a97463207d4a648f25a393be76a1da : SUCCESS in 5m 58s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2b05183352b45a9b0970dfb4a1c4845 : SUCCESS in 1h 52m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f3fd3c4a6fb4d72a5484bf43ffc29a0 : SUCCESS in 5m 58s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d8246f49b5a5428ab8fa60efe7a19eef : SUCCESS in 1h 25m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5dc9236e46264c23a9bf3dddab06f70c : SUCCESS in 1h 50m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/adfaf699e4ff4dd5bb2f9c9fe6a399af : SUCCESS in 32m 04s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/dc62ef88fe4144afbaefe68147cd8f67 : SUCCESS in 23m 13s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/60f97a4ec78649daa4ece540b6b12727 : SUCCESS in 7m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e948b4a97a984c1b88c8b3e52fd73cbb : SUCCESS in 1h 24m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e074a5f874a2237b78293e7b3910df346f460087","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-19 08:09:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3be134e4584f79dd90a52a8f9bdc3820d0e31f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-05-19 22:44:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"701d5b5a6437b8aed5fac39379a1d6528ba6d842","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-05-19 22:44:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"46e2dbdc7f7596bd2d64c8174e6dd2dee97a74e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 00:45:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d430e53dfa142399a2594d3356c7531\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c6529ee6ab71497991a40eb59b8b0dc8 : SUCCESS in 1m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c189bf35f7734cbc83d09d81203c2fec : SUCCESS in 6m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4514a17d53b459a98f27667b66ba02f : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b2ccb8a24c74b00ad62fee0621d9382 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56c1a292caa947e4833ba5f563e2d091 : SUCCESS in 2m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ed57483b884d462384b998b0ef0497d7 : SUCCESS in 2m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f66d0383e0634f378d10e9e5dd4c4cad : SUCCESS in 3m 55s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8330af05864c4071b259c77b8580fc7b : SUCCESS in 2h 01m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/681b8f4b4aa9490db266dfbe4e8690f6 : SUCCESS in 2m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f06ceb8ed76a47fba1df99e7577d6b5f : SUCCESS in 1h 23m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d0e165ca3b4e4468a228ad3cb5b972a5 : SUCCESS in 1h 44m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/167468892d2e4ccda00a21424bf423f5 : SUCCESS in 34m 17s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/b3eea279d04a47419f14c209fe80b4be : SUCCESS in 16m 20s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ba3e1a836a3347bfa0f241fa37e3acfb : SUCCESS in 3m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0df65e8ddec04113b69a95d05856247e : FAILURE in 27m 02s","accounts_in_message":[],"_revision_number":2},{"id":"dc6e81f8bf2fc67704d613e8767694413424869c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-07-07 04:37:38.000000000","message":"Abandoned\n\npbr is part of the API","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f943392305178ee4fa9444b9e70bd29049590aa3","revisions":{"da76c47b952e4216ec4723ac7ed49dabff2205f4":{"kind":"REWORK","_number":1,"created":"2026-05-18 18:54:01.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/19/989019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/19/989019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/19/989019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/19/989019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/19/989019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/19/989019/1"}}},"commit":{"parents":[{"commit":"6beaa5aa3a21cafa9eb1d23c49f0c9c966c4b48e","subject":"Merge \"Fix QinQ tag 0x88a8 in documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6beaa5aa3a21cafa9eb1d23c49f0c9c966c4b48e"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-05-18 18:53:05.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-05-18 18:53:53.000000000","tz":-300},"subject":"remove pbr as runtime dependency","message":"remove pbr as runtime dependency\n\nUse importlib.metadata to return the same version info instead of\ndepending on pbr at runtime.\n\nChange-Id: I044a1a3aaa92f9d05e68bd9a1d3d67a9ce3e2798\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/da76c47b952e4216ec4723ac7ed49dabff2205f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/da76c47b952e4216ec4723ac7ed49dabff2205f4"}]},"branch":"refs/heads/master"},"f943392305178ee4fa9444b9e70bd29049590aa3":{"kind":"REWORK","_number":2,"created":"2026-05-19 22:44:06.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/19/989019/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/19/989019/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/19/989019/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/19/989019/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/19/989019/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/19/989019/2"}}},"commit":{"parents":[{"commit":"6beaa5aa3a21cafa9eb1d23c49f0c9c966c4b48e","subject":"Merge \"Fix QinQ tag 0x88a8 in documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6beaa5aa3a21cafa9eb1d23c49f0c9c966c4b48e"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-05-18 18:53:05.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-05-19 22:43:54.000000000","tz":-300},"subject":"remove pbr as runtime dependency","message":"remove pbr as runtime dependency\n\nUse importlib.metadata to return the same version info instead of\ndepending on pbr at runtime.\n\nChange-Id: I044a1a3aaa92f9d05e68bd9a1d3d67a9ce3e2798\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f943392305178ee4fa9444b9e70bd29049590aa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f943392305178ee4fa9444b9e70bd29049590aa3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
