)]}'
{"id":"openstack%2Fneutron-lib~990237","triplet_id":"openstack%2Fneutron-lib~stable%2F2025.1~I77aea2b79a4321fafab64d1206169de604a89bc3","project":"openstack/neutron-lib","branch":"stable/2025.1","topic":"bug/2152898-stable/2025.1","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-08-04 07:24:18.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"36760":{"account":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"last_update":"2026-08-04 09:52:03.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-06-16 14:53:30.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I77aea2b79a4321fafab64d1206169de604a89bc3","subject":"Fix address‑groups API path and request body format","status":"MERGED","created":"2026-05-27 05:38:56.000000000","updated":"2026-08-04 09:53:00.000000000","submitted":"2026-08-04 09:52:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990237-bug/2152898-stable/2025.1","cherry_pick_of_change":988965,"cherry_pick_of_patch_set":1,"meta_rev_id":"1eea06d6b10580e1723010edf503291d8039d245","_number":990237,"virtual_id_number":990237,"owner":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-04 09:52:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2026-08-04 07:24:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2026-06-17 16:09:28.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2026-08-04 07:24:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-27 07:44:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-03 00:38:46.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-06-17 12:18:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"ded99e7d899d4b49f59496fab8e278e1edb189ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2026-05-27 05:38:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c213ac49040a7efa2ceb7833b7749dcbf6b4f2f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 07:44:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01080750ed304651ba0a5e0ac14db86e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6c9110028b844979a60abccaa46d518 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d097f0b949424cffb9cecd3ebb6a1ddb : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c4ea11015d349cbac990d769d76674f : SUCCESS in 3m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dbc5c818567e402ead53b393817e746d : SUCCESS in 1h 24m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/eb313e7cb45d44efac5ea2dc71f3335d : SUCCESS in 2h 04m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c8338179efbf4417a2488e0e53f152cf : SUCCESS in 34m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3e2de0490075462e939f4479aaaa39c5 : SUCCESS in 1h 35m 54s","accounts_in_message":[],"_revision_number":1},{"id":"41a89f77fd240f23522a1b46ca5d2035388f283e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-03 00:38:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6dd952fd7804e71ec1a8e7ad7d5954ad2f8c762","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2026-06-04 02:23:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51afbcc672dd14dfdf9f499212c3e2876fac2234","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-16 14:53:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ac9b7128ad2f837fd853ca3f155d03d31c5b219f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-17 12:18:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac38a0d71a795e53406e4a646ae2d1bef834807d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-17 16:09:28.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"074c25f10c4fc232d9ff7e71a80639920b141bc1","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2026-06-18 01:28:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9abe46d49cd3feefedb9ac7d3d62dfa1231efa80","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-04 07:24:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"18857f12c7ac9bc5f43673d2aa5714d8255e66f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 07:24:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"616be67b0cb0ec2c6dae68730dc242ddc7688152","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 09:52:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d44cd8e3769e415da6a328a6b6662b20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a9a21a0f065475392db3fdedd26059e : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be14d7381844ed68bb1f739b4718a03 : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7dbea857e5841f1b0bfd04e6e796a45 : SUCCESS in 2m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0abebdc7b3bc475ab1ec18abe5c66b11 : SUCCESS in 1h 23m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5333faad5423412fb3da430effa49437 : SUCCESS in 2h 26m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dbe8b4ef15c0460d91050b3d4da65e7b : SUCCESS in 1h 07m 07s","accounts_in_message":[],"_revision_number":1},{"id":"466b038aa8c2b04e56aa156be3145aead3c916c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 09:52:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1eea06d6b10580e1723010edf503291d8039d245","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 09:53:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16dd23f94aa643dfb4d8485d32a86bf2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93dc38346a6f4530a0cf598c6358587a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f63ad2e09f9494b47b72f5088860719536b3b94","revisions":{"6f63ad2e09f9494b47b72f5088860719536b3b94":{"kind":"REWORK","_number":1,"created":"2026-05-27 05:38:56.000000000","uploader":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"ref":"refs/changes/37/990237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/37/990237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/37/990237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/37/990237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/37/990237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/37/990237/1"}}},"commit":{"parents":[{"commit":"89473fdcb1e931c929572233a01194563c458d50","subject":"Merge \"Add the is_filter attribute to status on trunk port\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/89473fdcb1e931c929572233a01194563c458d50"}]}],"author":{"name":"Jimin Shin","email":"jimin3.shin@samsung.com","date":"2026-05-18 11:20:42.000000000","tz":540},"committer":{"name":"Jimin Shin","email":"jimin3.shin@samsung.com","date":"2026-05-27 05:38:56.000000000","tz":0},"subject":"Fix address‑groups API path and request body format","message":"Fix address‑groups API path and request body format\n\nThis change corrects the request path and body for the two\naddress-groups API endpoints listed below:\n1. /v2.0/address-group/{address_group_id}/add_addresses\n2. /v2.0/address-group/{address_group_id}/remove_addresses\n\nThe request path is changed from `/v2.0/address-group/...` to\n`/v2.0/address-groups/...` (note the trailing s).  The request body is\nchanged from { \"address_group\": { \"addresses\": [\"10.0.2.100/32\"] } } to\n{ \"addresses\": [\"10.0.2.101\"] }.\n\nCloses-Bug: #2152898\nChange-Id: I77aea2b79a4321fafab64d1206169de604a89bc3\nSigned-off-by: Jimin Shin \u003cjimin3.shin@samsung.com\u003e\n(cherry picked from commit 9122eb3d9fd0830f215122f3f2d578cac74ae309)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6f63ad2e09f9494b47b72f5088860719536b3b94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6f63ad2e09f9494b47b72f5088860719536b3b94"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"ours":"89473fdcb1e931c929572233a01194563c458d50","theirs":"9122eb3d9fd0830f215122f3f2d578cac74ae309","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
