)]}'
{"id":"openstack%2Fneutron-lib~990307","triplet_id":"openstack%2Fneutron-lib~master~I5986e8f1757067b470329d8b4cc9b4b9a84579a1","project":"openstack/neutron-lib","branch":"master","topic":"bug/2138746","attention_set":{},"removed_from_attention_set":{"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2026-05-28 13:29:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5986e8f1757067b470329d8b4cc9b4b9a84579a1","subject":"Trivial: Fix PVLAN_PLUGIN underscore","status":"MERGED","created":"2026-05-27 14:57:59.000000000","updated":"2026-05-28 13:30:37.000000000","submitted":"2026-05-28 13:29:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990307-bug/2138746","meta_rev_id":"4359b2044cb6047e1fc74b01c9a641cd60d5c6d8","_number":990307,"virtual_id_number":990307,"owner":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-28 13:29:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-05-28 09:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-05-28 11:34:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-05-28 11:34:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-27 17:04:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-28 09:28:31.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-05-28 11:34:51.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"68f49af61b77545f9d3fafbd7156a8b549bf4bbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-05-27 14:57:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcd4243fb972a7e30b85d1c898f4e1957b58cd32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 17:04:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ea2ee322d3047bdaa97629297ccfebf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42378ed09df941739663cfb4f085d1af : SUCCESS in 4m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bd75441a152418c999dd5c82c93ccda : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7616ccffe824214a198c52c06104bea : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ad50ce278ec43f5a8643e56900a9b7d : SUCCESS in 2m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ffcd3ebc7dc641d2b2bfbcd4a96355c0 : SUCCESS in 3m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/18a7b13cc94d4b899732de1ec02c78e5 : SUCCESS in 5m 23s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6213960f3f194c929b63389605353200 : SUCCESS in 1h 49m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3e9058ea30345c4bab3874914e990ec : SUCCESS in 3m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/51440c2ec65c4f2bb7317826804bd7dc : SUCCESS in 1h 26m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8e30758231a1460fb595147349c7a28f : SUCCESS in 2h 05m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/faae066582df49caa889e5f891582ede : SUCCESS in 32m 55s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/c2644300fd5a42fb88b85208205fe90f : SUCCESS in 17m 47s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3da2a2bece8d4f62bc8f99af306d7a73 : SUCCESS in 5m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0246c11ea9d6475b867502db4dde26ef : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":1},{"id":"8f789e0344b740765800b59706177ac5374c9dcb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-28 09:28:31.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0fe4e2b19ca5cdb1ad6b436ee2d59dc6c2b2a25","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-28 11:34:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2375321593a16f7b65f1327ec68dd0a90be5c2e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 11:35:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"661976884c0e2941a5e2e2c63d457bc71bfd38e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 13:29:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4485a8daeff14a818b434109abd4084e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b6ecaffb70a40e6b7e816ec1cc5b502 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81fffb6d35fb4ea697a45b5acbe16b08 : SUCCESS in 7m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2511384a8ac401188e758b67c11c9e8 : SUCCESS in 2m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4fbecd4a394d4868a27b6fea4795920a : SUCCESS in 3m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed90b1665efe4becb2d9632fbc3d4012 : SUCCESS in 1h 47m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b823622d88b42dd9c25d0338997b9e0 : SUCCESS in 4m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/627b8a574cdc4a5aaf4c0cb388f707b0 : SUCCESS in 1h 29m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bd293181fe9d48f9977db27ddf4831fd : SUCCESS in 1h 53m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/aaaf7ea9927b4f779b62f689a1f5b819 : SUCCESS in 1h 50m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0098fde0881b723cf811576734024969e31ec157","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 13:29:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4359b2044cb6047e1fc74b01c9a641cd60d5c6d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 13:30:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6871bfc6f87446fb4f7a58543d117ab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cd09f6d45d74e7a93374e15e7f77568 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8bbef6a5e40d3b8b71fe7add3a4cd3cd2156496b","revisions":{"8bbef6a5e40d3b8b71fe7add3a4cd3cd2156496b":{"kind":"REWORK","_number":1,"created":"2026-05-27 14:57:59.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/07/990307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/07/990307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/07/990307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/07/990307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/07/990307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/07/990307/1"}}},"commit":{"parents":[{"commit":"07d282c919a38fafe0ff3c90c6ce7d7b9f3dd0a4","subject":"Merge \"Fix address‑groups API path and request body format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/07d282c919a38fafe0ff3c90c6ce7d7b9f3dd0a4"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-05-27 14:56:26.000000000","tz":120},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-05-27 14:57:23.000000000","tz":120},"subject":"Trivial: Fix PVLAN_PLUGIN underscore","message":"Trivial: Fix PVLAN_PLUGIN underscore\n\nTrivialFix\n\nChange-Id: I5986e8f1757067b470329d8b4cc9b4b9a84579a1\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8bbef6a5e40d3b8b71fe7add3a4cd3cd2156496b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8bbef6a5e40d3b8b71fe7add3a4cd3cd2156496b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
