)]}'
{"id":"openstack%2Fneutron-lib~990447","triplet_id":"openstack%2Fneutron-lib~master~I202d16b829f43f089893d23e8ae0b901a7b58e3f","project":"openstack/neutron-lib","branch":"master","topic":"bug/2138746","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-05-29 14:55:46.000000000","reason":"Change was submitted"},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2026-05-29 14:55:46.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-05-28 22:32:14.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I202d16b829f43f089893d23e8ae0b901a7b58e3f","subject":"Change PVLAN_PLUGIN location","status":"MERGED","created":"2026-05-28 11:40:55.000000000","updated":"2026-05-29 14:56:52.000000000","submitted":"2026-05-29 14:55:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990447-bug/2138746","meta_rev_id":"6ffbb9b6ad8c22039b68418a12f0f126d86076c4","_number":990447,"virtual_id_number":990447,"owner":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-29 14:55:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-05-29 12:02:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-05-28 22:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-05-29 12:02:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-28 11:43:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-05-28 14:01:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-28 17:41:19.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-05-28 22:32:14.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-05-29 12:02:12.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"30527a38c0bf1fe55b5999ebde99f2ea491ec71b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-05-28 11:40:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6b9e288a193e8d1ea484b09bdf53cf87acde3d3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-28 11:43:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0eb391e24eb587cf831832d15a3cc29b25dc6d51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 14:01:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce75fa8d1cce4d48b9ef980bc228d855\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f0ee85fce346a896cc7532ae05c4d1 : SUCCESS in 8m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/985d365cfc814a35ba54fb0e3ea6c316 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c16bc88094c42e6b59805777a5baee6 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21ec4ea1989942139ec0d8c2967649f4 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e02d39694cb447bb4342948d756c72c : SUCCESS in 3m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f024050ec710453bb7266e2fa515fd3f : SUCCESS in 5m 54s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62e57eb19f5b4b08932eeeb7990d1d30 : SUCCESS in 1h 59m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c83a377f6f6a46d484cda51d5de0d4f6 : SUCCESS in 2m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/836b95bc1b2640d69f1763ab28533da8 : SUCCESS in 1h 45m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/831b4f9b0e3e4711a694a05160766135 : SUCCESS in 2h 19m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e69b55fd184745abbcce3bc1ba63f623 : SUCCESS in 35m 18s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/816e1d5f502d49dd8dc8f2df00d69a3d : SUCCESS in 25m 26s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8e6b28c25db44e49834e0e16ed2e00d0 : SUCCESS in 5m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0616b4e4a8114e53ada2d797f539ce34 : SUCCESS in 1h 38m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d959f52f94e7676503e01e4a8ff871a09ca93e88","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-28 17:41:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc37cf20eb11801fafaa44eecac4d5e040f0d5e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-05-28 21:08:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"78ebace446cd0b9a6e6d619f85ef2cbe70b0e97e","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-05-28 21:09:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f06376b40af5b3bc5521f0add0d5bf3771743c55","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-28 22:32:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"04ca7365f8b6724bc4271fbe986f78ea8dc2c878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 22:50:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96ce99afd9324381942a6e9a2a9560a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e151b6d43ab146d299d8a4aed9996f8a : SUCCESS in 9m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc79e11c0f1c476b964b9fd7c4d85665 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5fb3527cd974340b1106efeaf964b3b : SUCCESS in 7m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26bd74702060438c874c3bf5d7d9f212 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ab20f81af4e4420979c8f17438305be : SUCCESS in 3m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c7cf8fedc2134e44b54ea30ecc0bbced : SUCCESS in 5m 50s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0698e1211b35488d80d19249c4597b60 : SUCCESS in 58m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e50f0bc853042d886865f045ce20564 : SUCCESS in 4m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ae57a0467e60457e921afa969753abf3 : SUCCESS in 1h 23m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d7946fdf0bdf4a3c85b6717e7fbe5bcd : SUCCESS in 1h 40m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9ef7eff0700844e98c4b3faabd659178 : SUCCESS in 32m 38s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/2d8199d0de9c4a98824bdd854c44f573 : SUCCESS in 18m 47s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/4cefc8a236a840a0ad2c24851d2c4b58 : SUCCESS in 5m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/149c9ca5d9dd460db266cf9f73b9f1f4 : SUCCESS in 1h 24m 34s","accounts_in_message":[],"_revision_number":2},{"id":"27db228c24c9de47d900b2bce957143bccba2383","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-29 12:02:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f4922e342315c8be44c643d70cf51e609b8ac1f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 12:02:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6129bc5a44b87f41ce53fc42751f0027f237b1f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 14:55:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/677891f42df140898aa95769dd5c61c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3549c15ae5104409b251da3f488c3f65 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d29ff4e29ab04f739c431cc6e011d5dd : SUCCESS in 8m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ac27f71d2214b1da2dfcef3541b27b6 : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7973a4d12964864a1971453e27901c9 : SUCCESS in 2m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/658172db9ecc45babca20cedba36da1d : SUCCESS in 1h 44m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c06c43b8294c413089434c8108134f63 : SUCCESS in 3m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3d9f629a655a4ec79b91da756459c456 : SUCCESS in 1h 30m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7180dc7c152e4e968b68fc21fdaec863 : SUCCESS in 2h 52m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ed85295a13324112979a0c9fc336d9c6 : SUCCESS in 1h 25m 18s","accounts_in_message":[],"_revision_number":2},{"id":"8fb1793aa252aeed47ccdc458215b843185bc7b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 14:55:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6ffbb9b6ad8c22039b68418a12f0f126d86076c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 14:56:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f8b53c3d5f04619b25c6226a5728dd3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53f3e3e220fa4215a00ea345bd7eeb36 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b64956323e9d2c72412c4a374dfc1b00f5200590","revisions":{"0e9f1f554bd152dbd612b6f728df25d6644f4fe3":{"kind":"REWORK","_number":1,"created":"2026-05-28 11:40:55.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/47/990447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/47/990447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/47/990447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/47/990447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/47/990447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/47/990447/1"}}},"commit":{"parents":[{"commit":"8bbef6a5e40d3b8b71fe7add3a4cd3cd2156496b","subject":"Trivial: Fix PVLAN_PLUGIN underscore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8bbef6a5e40d3b8b71fe7add3a4cd3cd2156496b"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-05-28 11:37:47.000000000","tz":120},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-05-28 11:40:30.000000000","tz":120},"subject":"Change PVLAN_PLUGIN location","message":"Change PVLAN_PLUGIN location\n\nSince we are using PVLAN_PLUGIN constant in events, it is better if we\ndefine it in callbacks folder instead of the dedicated PVLAN constant\nfolder.\n\nRelated-Bug: #2138746\n\nChange-Id: I202d16b829f43f089893d23e8ae0b901a7b58e3f\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0e9f1f554bd152dbd612b6f728df25d6644f4fe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0e9f1f554bd152dbd612b6f728df25d6644f4fe3"}]},"branch":"refs/heads/master"},"b64956323e9d2c72412c4a374dfc1b00f5200590":{"kind":"REWORK","_number":2,"created":"2026-05-28 21:08:35.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/47/990447/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/47/990447/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/47/990447/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/47/990447/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/47/990447/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/47/990447/2"}}},"commit":{"parents":[{"commit":"8bbef6a5e40d3b8b71fe7add3a4cd3cd2156496b","subject":"Trivial: Fix PVLAN_PLUGIN underscore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8bbef6a5e40d3b8b71fe7add3a4cd3cd2156496b"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-05-28 11:37:47.000000000","tz":120},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-05-28 21:08:28.000000000","tz":120},"subject":"Change PVLAN_PLUGIN location","message":"Change PVLAN_PLUGIN location\n\nSince we are using PVLAN_PLUGIN constant in events, it is better if we\ndefine it in callbacks folder instead of the dedicated PVLAN constant\nfolder.\n\nRelated-Bug: #2138746\n\nChange-Id: I202d16b829f43f089893d23e8ae0b901a7b58e3f\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b64956323e9d2c72412c4a374dfc1b00f5200590"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b64956323e9d2c72412c4a374dfc1b00f5200590"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
