)]}'
{"id":"openstack%2Fneutron-lib~991527","triplet_id":"openstack%2Fneutron-lib~stable%2F2025.2~I77aea2b79a4321fafab64d1206169de604a89bc3","project":"openstack/neutron-lib","branch":"stable/2025.2","topic":"bug/2152898-stable/2025.2","attention_set":{},"removed_from_attention_set":{"36760":{"account":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"last_update":"2026-06-25 11:01:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I77aea2b79a4321fafab64d1206169de604a89bc3","subject":"Fix address‑groups API path and request body format","status":"MERGED","created":"2026-06-04 02:22:44.000000000","updated":"2026-06-25 11:03:09.000000000","submitted":"2026-06-25 11:01:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991527-bug/2152898-stable/2025.2","cherry_pick_of_change":988965,"cherry_pick_of_patch_set":1,"meta_rev_id":"dcedf12d30d286e7932a5e480ec05cb3d24825f4","_number":991527,"virtual_id_number":991527,"owner":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-25 11:01:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-06-25 08:36:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-06-16 14:53:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2026-06-25 08:36:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-04 04:59:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-16 14:53:13.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-06-25 08:36:29.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"fdb6222d2690d5486e8cb967dcc3fcf970de98de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2026-06-04 02:22:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8756416c3ec1c708ba37ad607ab0eb86ce14f580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 04:59:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0dd20f83b0a842fa9bb9c7f0538be1e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96173a72b3dc4ac09dbef9356f9c41c1 : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96c01ad2de9e41fb8829d2602ec6bc8c : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d3663eb4ddb42c4899238c3677d8108 : SUCCESS in 5m 48s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/64785dfdb30f4f7c9b3766c2f3601324 : FAILURE in 8m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/6c0544a79bc140568d45767dfbcdd74c : SUCCESS in 1h 45m 09s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/da2faae8e84d40f6a464c873ad474f21 : SUCCESS in 2h 11m 58s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/afec62a6becf4d82948c59c8baa80198 : SUCCESS in 31m 56s","accounts_in_message":[],"_revision_number":1},{"id":"0bb9a14943e38a27ea2cb6a287d23a044a8d1f99","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2026-06-08 06:05:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d054d0e08fe0cf6f6cef3148c0872c5e36627e62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 07:53:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4156a73e302429494806da5ac29edf5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/308db5b3ab2d4711a8cfb0fce57aee8c : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c45b992b97d4f84ae37a04489de688f : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57eddc680825479bbe964c5f4c552882 : SUCCESS in 5m 27s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/5f70d03cfbb74ed3aca6cccf61a4e916 : SUCCESS in 1h 29m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/1c5c0fd8c106479f8bdc7d42f14229c7 : SUCCESS in 1h 35m 54s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/972568830ba64388ab8a5555df272f79 : SUCCESS in 1h 46m 38s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/42f6bff45c66498aabf129f8a9b14d07 : SUCCESS in 34m 45s","accounts_in_message":[],"_revision_number":1},{"id":"667944c76c4ebb484cbebe77ceb31b27e637846b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-16 14:53:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"586e4fb16e188fcad9f26c0b4ecd2a66646b2cd8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-25 08:36:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c33d3a48a04ace2fc12e08a9a2593bdfbb519f71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 08:37:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c2e2242b4c7cfff26977f6ff783bbcc2d727ad5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 11:01:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e312d0106ff544858059ec8f6c2d7351\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/490c216e86e545ee95c06e165cd8cb87 : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e20a23c9c0134ef5a1d13e945a83daeb : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdf9309c54734148b6050c41aaaaa34a : SUCCESS in 5m 04s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/a4ff94684cab44739fdc0b45fcf55a9b : SUCCESS in 1h 42m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d27f4882b52ab67403bf636fe88a0afee24a5108","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 11:01:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dcedf12d30d286e7932a5e480ec05cb3d24825f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 11:03:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ea27d663209419e890a4bd34abcbf56\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25b35beb8aa74741aa0ec4e92ed7d660 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0fbd726e9df7b9918c636c82802a4e9d9f07635","revisions":{"f0fbd726e9df7b9918c636c82802a4e9d9f07635":{"kind":"REWORK","_number":1,"created":"2026-06-04 02:22:44.000000000","uploader":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"ref":"refs/changes/27/991527/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/27/991527/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/991527/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/991527/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/991527/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/27/991527/1"}}},"commit":{"parents":[{"commit":"bbfc12263835e7d6e07c74b26a1bc5a46e304633","subject":"Merge \"Add the is_filter attribute to status on trunk port\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bbfc12263835e7d6e07c74b26a1bc5a46e304633"}]}],"author":{"name":"Jimin Shin","email":"jimin3.shin@samsung.com","date":"2026-05-18 11:20:42.000000000","tz":540},"committer":{"name":"Jimin Shin","email":"jimin3.shin@samsung.com","date":"2026-06-04 02:22:44.000000000","tz":0},"subject":"Fix address‑groups API path and request body format","message":"Fix address‑groups API path and request body format\n\nThis change corrects the request path and body for the two\naddress-groups API endpoints listed below:\n1. /v2.0/address-group/{address_group_id}/add_addresses\n2. /v2.0/address-group/{address_group_id}/remove_addresses\n\nThe request path is changed from `/v2.0/address-group/...` to\n`/v2.0/address-groups/...` (note the trailing s).  The request body is\nchanged from { \"address_group\": { \"addresses\": [\"10.0.2.100/32\"] } } to\n{ \"addresses\": [\"10.0.2.101\"] }.\n\nCloses-Bug: #2152898\nChange-Id: I77aea2b79a4321fafab64d1206169de604a89bc3\nSigned-off-by: Jimin Shin \u003cjimin3.shin@samsung.com\u003e\n(cherry picked from commit 9122eb3d9fd0830f215122f3f2d578cac74ae309)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f0fbd726e9df7b9918c636c82802a4e9d9f07635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f0fbd726e9df7b9918c636c82802a4e9d9f07635"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"ours":"bbfc12263835e7d6e07c74b26a1bc5a46e304633","theirs":"9122eb3d9fd0830f215122f3f2d578cac74ae309","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
