)]}'
{"id":"openstack%2Fneutron-lib~991962","triplet_id":"openstack%2Fneutron-lib~master~Ia374921f7c713561ba9f818b8456594cea3fc8e2","project":"openstack/neutron-lib","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-09 13:24:44.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ia374921f7c713561ba9f818b8456594cea3fc8e2","subject":"Run mypy from tox","status":"MERGED","created":"2026-06-06 10:45:38.000000000","updated":"2026-06-09 17:31:30.000000000","submitted":"2026-06-09 17:30:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991962-typing","meta_rev_id":"fcf8b6a2702e0388b347ce7cae007ed1d45ecf79","_number":991962,"virtual_id_number":991962,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-09 17:30:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-08 12:49:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-06-08 14:52:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-08 14:52:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-06-08 12:49:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-06 12:40:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-08 12:49:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-08 14:52:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"4bbbcc4f3923865a74180b9ede769b72f75e11b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-06 10:45:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"146f84e03e1f83ecf9ccd9c8bffd86ffc7e5e203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 12:40:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/502b24e1c7b94d0c8d955927ae059dc2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2480c63d3b7e4b7481151081e41c781b : SUCCESS in 3m 42s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e0d4648e069c472a97590fd328bd69c0 : SUCCESS in 2m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a96483c91ec640b99259bbb11aa5a70e : SUCCESS in 1m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7f71d9f4b3b42d982405ee3e12ee601 : SUCCESS in 8m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/399ac8cd8de14935ad91d4faaed3764b : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f76e89585b84b4f9500b85d3e3b665b : SUCCESS in 6m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c540ecfcf6df485fb67a7b23fdb38432 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3e9a5b786744b71af4bcca65cc6e7e3 : SUCCESS in 3m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f04fc899d10544dc8b45db3b44085fa0 : SUCCESS in 5m 12s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29fa90bf0e614e4b8f657f6868ae6f1f : SUCCESS in 1h 41m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/797730949cb245fb8cac444de76ae254 : SUCCESS in 5m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cd80d593c9d749efa45717e327271700 : SUCCESS in 1h 21m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/37710c7870f3446880ce35a51f0b0596 : SUCCESS in 1h 53m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d56b2b7117d847d6a943d9a95826a63e : SUCCESS in 33m 51s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/2e215d8161744426a310fdba627bd72a : SUCCESS in 24m 38s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d09cf4f4efdb4f1aaaa2aeb2194a6786 : SUCCESS in 3m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e15ae7b1a94740c9810deeed17700475 : FAILURE in 50m 55s","accounts_in_message":[],"_revision_number":1},{"id":"55433fd652970f34d48bc899542d047c00540e47","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-08 12:17:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6dd571d537adbb79f1684fe62c8fbfc50f79bf2a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-08 12:49:18.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"ec6a462e8539405cb84f9ec2c8f911f2c498b121","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 14:13:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f68c92b02b50421fa449ee0cd70bb9fa\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c6d069f22e8e4200b5b27aa9ae16b1ee : SUCCESS in 4m 16s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b164638995bc4083ab69500ec533195b : SUCCESS in 2m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4595584548104b7ca26aa4f40fa87917 : SUCCESS in 1m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4d227952ab6473db35a147714d7c765 : SUCCESS in 8m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e71aecd67574a338c2e19c65de83711 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b9c6d8c5fb943b38eeef666ab8fb978 : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/115476763a8446e28d998700cb26b5b9 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86af9834d00e4fb09a4a23f49b5f3877 : SUCCESS in 3m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b1cd2809ee11492193143c1d84384735 : SUCCESS in 8m 11s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7476eed0fda244e88f88ab8cfd98da4b : SUCCESS in 1h 31m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/454752e9721a43ec8396cc95fb2b738c : SUCCESS in 4m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b2feab8024bb4d62ab60cc4698b2156e : SUCCESS in 1h 17m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cca01fea2b45429d9aab7c92caf065e5 : SUCCESS in 1h 54m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/873d1cd7203549669ed6db4b3f74944f : SUCCESS in 32m 34s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/19424f2041244192ab26f7d420e180bb : SUCCESS in 22m 59s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9bee1a0256c348778999e7527d09f493 : SUCCESS in 5m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/08495e985058408189fe5e99e707ab56 : SUCCESS in 1h 24m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ddb023fac3f2f98d882754b4782aff0e72ebe4b2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-08 14:52:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ebc4270f6856dc38bf19ef3ea09952afd4c5052","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 14:54:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f1acfe0b1e21cdb339931632647d3e69875b34f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 17:11:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/10680499a51c4b96983e5dca3177cea6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9908999baeb24841a19c9d79d70863ed : SUCCESS in 3m 51s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2af074be82e743d298d4c3938632e2ef : SUCCESS in 2m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3b016a344a4b41e6b074eb49e546c675 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e5650ca5e39440bb6afe1db37aeeb39 : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78c1a09c2a1345159d859c7ede427f77 : SUCCESS in 8m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bce89fe2fe4949799d3b094c897194d5 : SUCCESS in 6m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/868020c48b924a66ae370b252c99eab8 : SUCCESS in 3m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83dc4f4370db4c848f5c53fd0734dd08 : SUCCESS in 1h 43m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ec17b83600248ba9aad92958d02ea82 : SUCCESS in 5m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/69b74d1c929447d0a7f751e8868193cc : SUCCESS in 1h 36m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cfaecfc061444cdc83431fe2ba301d61 : SUCCESS in 2h 15m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3b776b3034b7413ebaba2e598ff27da1 : FAILURE in 1h 22m 30s","accounts_in_message":[],"_revision_number":1},{"id":"bbfaf9db6a0a307620b970e762f8e990c92a767b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-09 06:56:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4df750717205151343c0ad69466606104b62a82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 09:33:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44eae580324041b1b1ed8425e5210aee\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5cfac77468144a07a3b04eccf6299bc1 : SUCCESS in 3m 38s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fe101a46a7604f8da733500476ae658e : SUCCESS in 2m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/33bea1b9f1d147caac0c11a38a6f25aa : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78307ac5c4d14ffcb131a95b80dea15a : SUCCESS in 7m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8024774d5d7d4664b32f3acb3fb3830f : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd21d57b11b0494887651a2d28afedee : SUCCESS in 6m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60752853465c4dcabbde390b335bba7a : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cfced24db2e44d82a11658c8a8c950e6 : SUCCESS in 2m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/22770fc1297d42759e0f9a9f8fc53f72 : SUCCESS in 5m 52s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bae5534871614520999b8eb2bc555308 : SUCCESS in 1h 43m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9278f8d956364459a69c23b5c48c9873 : SUCCESS in 2m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c4ff723713284ab9b9bf3fe7e955b1e6 : SUCCESS in 1h 34m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ba45495f397045efb6d881f5a7a1f677 : SUCCESS in 2h 35m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/088f164e80d74db19c2557ccbecaadd3 : SUCCESS in 35m 31s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/379cdd2877a34a4e931604eccbcd483b : SUCCESS in 21m 07s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ad439f8f2f70423a9aea6041f1017afd : SUCCESS in 5m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b2413a01f457465b977fd8899f01c954 : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":1},{"id":"bf01e18c8e8bbb76bd05c68cfcadde726e92074d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 09:33:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"de1ea2e9ee771075b85016cce70a0f52ad5c4e0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 11:48:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/636a450cbce04d1e9eb08da997c6dda6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/af308ae3ef92442c8d93c6fc9da92e79 : SUCCESS in 4m 30s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/96baa117b37f4fc89ddfa2998174eea4 : SUCCESS in 1m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ec0431ee64b54b39a938fd2b63256aab : SUCCESS in 2m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c85c92270dd472b9007ec1503ef3d6a : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/191d658dc8504205af6db0d2d6fed0c2 : SUCCESS in 7m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e7c25d523b08434f82865c1a45f95087 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/801c829f1db04cac99beab765e0c820b : SUCCESS in 4m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8caa6165821549c189f0012c99193f29 : SUCCESS in 1h 03m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89ece51c0384433ca9b38a2494274063 : SUCCESS in 4m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/78595064c77b4f508ae35cb174ba39c2 : SUCCESS in 1h 18m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/18ee3151bfd24728b064635e2142f289 : FAILURE in 2h 12m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/63e156bd6ba744dd97267f14ae4e92bf : SUCCESS in 1h 33m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c699722012084cad7287b92959e6cbff757ca2be","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-09 13:24:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dab2a90a296df8bcbcf89f5aba73a6d29541bc91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 15:32:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91b4efee00c9479787a4dfea4052a4be\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/deb4479cdbb14e83a713067638df2952 : SUCCESS in 4m 00s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f55370df9fe443c5b50e5e79ae004cf3 : SUCCESS in 4m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/de5b3e6ba9994be3a4be228b89b6a16e : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512637b190b149c58ca32f7b28a14e08 : SUCCESS in 9m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82c5b3fddb8347af9530a3f28f1ad9aa : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89f8413c15be480bb1e32de969af9b09 : SUCCESS in 9m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d249070c65c44e049eef3c13c7d24468 : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0828e1c827f64b4e91871d168a851167 : SUCCESS in 2m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/002f7721292740d3971f8da76cb64011 : SUCCESS in 5m 16s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0308d60f96cd4de8bb698582b77b51e0 : SUCCESS in 1h 56m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/357673129ed84c4d852fcf9ec124fcc0 : SUCCESS in 5m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ad19b53e389940aeb87db32376a87861 : SUCCESS in 1h 19m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a66efa9919494439bb5b3a149cb1bfa8 : SUCCESS in 2h 06m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2b8d5103cfb64709a0be938284840d14 : SUCCESS in 34m 10s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/cdd51f7ffe194b97a4078ff557cf7aee : SUCCESS in 29m 57s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/01036e79410f43d3985d273223a23a4d : SUCCESS in 5m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/20fc904b20bd4d0f8ec2420f5dffc25c : SUCCESS in 46m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d4f9fb4a7b68eed3dd918433a3cf7e329a8070c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 15:33:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1077c1baac19b44c32cb7802bcdb629dbe7f3dc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 17:30:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09c0988a4d18450fba820c1193b9cf83\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/50950a9c425f4fee95fafaab024bffe8 : SUCCESS in 3m 34s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c94b6b85d3cb48a89dc22c682c7adb68 : SUCCESS in 2m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/92a201c1a99045bbad02961da7633d1e : SUCCESS in 2m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f07ba2708784733af6ba534cbeb9e46 : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6074a085a0204f37b9a54b61f824aada : SUCCESS in 6m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa033e42a7b24e62999c46d369c33acb : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3974e4d3311e4067aac8e75cf0f8e06c : SUCCESS in 3m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa5585796c6f448fb46547793c6d5242 : SUCCESS in 1h 55m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/612e626e0690455890d51f83ed8c62e4 : SUCCESS in 3m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0bdcb44e1cf1426d90895a1223904db8 : SUCCESS in 1h 25m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bd9a7048b0a14dd5be1d6e5619e0d855 : SUCCESS in 1h 42m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f70b1c8db1314c7b865f91788b8479d7 : SUCCESS in 1h 07m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a5308e6e10d855681a8a3628b3bd2d8eb3f7fe21","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 17:30:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fcf8b6a2702e0388b347ce7cae007ed1d45ecf79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 17:31:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b91488e19c248a596ea323b9f7ab7b6\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/e8a532028e68441aa9ff1c43c7d671b7 : SUCCESS in 1m 02s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f20df0753f6f413998a85c83068ed57b : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f6156b39e2d46ea87356740814e7189 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc6c8e24e32398e681f775ddd529d4e233befb3f","revisions":{"cc6c8e24e32398e681f775ddd529d4e233befb3f":{"kind":"REWORK","_number":1,"created":"2026-06-06 10:45:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/62/991962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/62/991962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/991962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/991962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/62/991962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/62/991962/1"}}},"commit":{"parents":[{"commit":"8788440009b5b545265797ce016d8d2a69a83efe","subject":"objects: Correct ListOfDictOfMiscValuesField.AUTO_TYPE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8788440009b5b545265797ce016d8d2a69a83efe"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-02 16:48:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-06 10:13:53.000000000","tz":60},"subject":"Run mypy from tox","message":"Run mypy from tox\n\nThis avoids the need to duplicate our dependency list in multiple places\nand allows us to take advantage of tox\u0027s dependency management\ninfrastructure, to ensure we always get the latest and greatest version\nof a package allowed by upper-constraints. It also removes the slowest\npre-commit hook (by some distance), making pre-commit faster\n\nChange-Id: Ia374921f7c713561ba9f818b8456594cea3fc8e2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cc6c8e24e32398e681f775ddd529d4e233befb3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cc6c8e24e32398e681f775ddd529d4e233befb3f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
