)]}'
{"id":"openstack%2Fneutron-lib~991964","triplet_id":"openstack%2Fneutron-lib~master~Ib0ea320608cb30e5b094670100ed398adc6e5f74","project":"openstack/neutron-lib","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-17 16:32:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib0ea320608cb30e5b094670100ed398adc6e5f74","subject":"typing: Add types for SUB_RESOURCE_ATTRIBUTE_MAP","status":"MERGED","created":"2026-06-06 10:45:38.000000000","updated":"2026-06-17 16:34:03.000000000","submitted":"2026-06-17 16:32:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991964-typing","meta_rev_id":"3af0bfc47999f7445772cf5ca3aa933e73865f5d","_number":991964,"virtual_id_number":991964,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-17 16:32:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-17 06:39:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-06-17 14:24:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2026-06-17 14:24:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-06 12:39:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-08 12:50:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-06-16 13:10:26.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-06-17 06:39:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-17 14:24:58.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"85cf263181546092b91f93606000af7b17a09277","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-06 10:45:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d1e82013d2bc5ac06c55db8609bc659c985b3b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 12:39:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7c10034ffcc4429810b91ab946c6f9d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36b6380c916b4479a99733adb8bb52c8 : FAILURE in 6m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ef6a56168d4439688950b915df98874 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7086faa262844bc88a1906b64dc114c : SUCCESS in 7m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec238c1a31c34fada24b2cf95fa1dc51 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3e6ae15befa6485291e7f9e0ce374ac8 : SUCCESS in 2m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2e37f36ddab849cebca57169f41a03bf : SUCCESS in 5m 58s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/610c5690a9db470a95d2771fc23249be : SUCCESS in 1h 48m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d3d0691d35b4f30ad90cbc74b982079 : SUCCESS in 2m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f753b5816f314583a9d068ce7a97668c : SUCCESS in 1h 22m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4fbfe5cdd33440a793f4750c69547173 : SUCCESS in 1h 46m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4debabb0adf5405ea0455c8dec24ab8b : SUCCESS in 32m 10s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/908e8fd927a34d519f44ffc58ded3dae : SUCCESS in 21m 51s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e4895b2902c847628b4bff83238a873a : SUCCESS in 5m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8e09f70a2cc0439aab23a210b3d5fe32 : FAILURE in 27m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bd92b53766ad383e02d79c9da542bebc87b21ab9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-08 12:27:24.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"6453c22b6827a2a0ebf61ae69440d1b7287b1e75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 14:34:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd18503031c541f0b744829bea4c8fef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd80d1aacb2f44ccadf54d5929d5b3b2 : SUCCESS in 7m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5765829319d74e20beb610dc97d27bf6 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cdcb8237ffb47588218a65b5256a743 : SUCCESS in 6m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/792b2930e51c41699cb4a2d46b18a73b : SUCCESS in 2m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/33edd296dc7c4c678c1688921c77c0e9 : SUCCESS in 3m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b7b79a316b774664a280e63593332878 : SUCCESS in 6m 19s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e575844ab06047d5b5a80ce72da125c1 : SUCCESS in 57m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94fa242620124e8fb2e296f8932ed327 : SUCCESS in 5m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c5deb05163b34b3698ae3f3308e247c2 : SUCCESS in 1h 26m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bdf4f4d07136451ca2c0e52bbc4f200d : SUCCESS in 2h 04m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8befca0c0efa4500af097ef18abe336f : SUCCESS in 32m 11s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/721ba3f162d84d25815f7976f54ee472 : SUCCESS in 19m 01s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7065e776df2344319a4d8ddf7055f413 : SUCCESS in 5m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/44435defdde9439ba97828e877c6977c : SUCCESS in 1h 35m 19s","accounts_in_message":[],"_revision_number":2},{"id":"9f78066ff50488ebc8ab688b7cb93b32d3887a1f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-16 13:10:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"21c34fa77be19fd87f411f8ea72b3b5e48d1c02e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-16 13:14:19.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"b6c927d180c8c9829a17f30222d27faa3c6e9db9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-16 15:58:45.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f5c53f24c262e4ff1c58e1789fc164f9faa53c78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:36:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6d20b3824184f03a6c64ce730cc2bb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed5b562b5c054f3c84f95d5ff9cf54d1 : SUCCESS in 9m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e848b0eecbce4e15bc0553c64071123b : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d13e538b76b4bd2a2db240302d39774 : SUCCESS in 7m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/648c87b4c12f48d4b1296742399c34e7 : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a0d6d1789a34e638b9cd2eb25ad4f30 : SUCCESS in 3m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6d94b6bef7254709b616cd90718b5eb9 : SUCCESS in 7m 21s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f7c7786ea4f4048b44b85ca3857c0c1 : SUCCESS in 58m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0688c61cc2fd47afa26c390b39c21d30 : SUCCESS in 5m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5b0ee0d87b6049f0ac7c3c280c61a9bb : SUCCESS in 1h 22m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f709697e1cd5462f8e1f89f5a67fcbcb : SUCCESS in 2h 33m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2f1c15fb0d0c42d496b59ff2c0cada91 : SUCCESS in 31m 21s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/e15d43877a0345b9845ed7dc39b67a02 : SUCCESS in 25m 03s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/90ddb3ab59e34dfab2566c014223c27e : SUCCESS in 4m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dfa9e7b473814d9cac3deb96c5121931 : SUCCESS in 2h 02m 32s","accounts_in_message":[],"_revision_number":3},{"id":"59fc118508c1591862f61659e1fff86ae7b275ca","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-17 06:39:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"05409ba34474d8f3f6ce64f36c756a291bc60f34","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-17 14:24:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"81b9f5e5bc9a52654485f8fc41796aa749ec2a90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 14:25:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"20902d553f078ea31e606bbe52dffea415becade","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 16:32:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/935aeb862aa94a1c81b8032f9b4aefa8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/861047ea7e77408c817d78f8c83e52d7 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce15059bca9544d19f793cfefd212849 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17e6203772f94db89f29527c6c909981 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1451fc7b3660407baef315c796b34cd6 : SUCCESS in 3m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34857cfc4b9e4d0886de179fa42855a9 : SUCCESS in 1h 46m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1845e643c824bcfaf6be8d9ff5c1de8 : SUCCESS in 3m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7f050f755ad14d5da627ca33ae3f8890 : SUCCESS in 1h 55m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0dab3ee95ba441489fa484c14610296a : SUCCESS in 2h 06m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/12d5bd405370435a8cc935d754c13b72 : SUCCESS in 48m 12s","accounts_in_message":[],"_revision_number":3},{"id":"edafd62bc7a0abd9488d0b805d792666da1e7d87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 16:32:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3af0bfc47999f7445772cf5ca3aa933e73865f5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 16:34:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ba728cc103b48169ab048cac4ac9ef7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc236d9ee6b1497ab033bffa78b29f51 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b8010afeb2e2954b84f0cdf8ebfdb5ecb3da7221","revisions":{"89811ccb9b94dfff914b636db2c7950b86aacf39":{"kind":"REWORK","_number":1,"created":"2026-06-06 10:45:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/991964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/64/991964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/64/991964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/64/991964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/64/991964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/64/991964/1"}}},"commit":{"parents":[{"commit":"c49fd3ec99873b3b9281a4a702ff801a6447edb8","subject":"typing: Add types for RESOURCE_ATTRIBUTE_MAP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c49fd3ec99873b3b9281a4a702ff801a6447edb8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-03 13:37:09.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-06 10:13:53.000000000","tz":60},"subject":"typing: Add types for SUB_RESOURCE_ATTRIBUTE_MAP","message":"typing: Add types for SUB_RESOURCE_ATTRIBUTE_MAP\n\nChange-Id: Ib0ea320608cb30e5b094670100ed398adc6e5f74\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/89811ccb9b94dfff914b636db2c7950b86aacf39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/89811ccb9b94dfff914b636db2c7950b86aacf39"}]},"branch":"refs/heads/master"},"dd12839743092f3a1026d763346180473e2788d7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-08 12:27:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/991964/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/64/991964/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/64/991964/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/64/991964/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/64/991964/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/64/991964/2"}}},"commit":{"parents":[{"commit":"2b44362aa27acea5b01871614e8e72cc72cafa17","subject":"typing: Add types for RESOURCE_ATTRIBUTE_MAP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2b44362aa27acea5b01871614e8e72cc72cafa17"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-03 13:37:09.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-08 12:27:10.000000000","tz":60},"subject":"typing: Add types for SUB_RESOURCE_ATTRIBUTE_MAP","message":"typing: Add types for SUB_RESOURCE_ATTRIBUTE_MAP\n\nChange-Id: Ib0ea320608cb30e5b094670100ed398adc6e5f74\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dd12839743092f3a1026d763346180473e2788d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dd12839743092f3a1026d763346180473e2788d7"}]},"branch":"refs/heads/master"},"b8010afeb2e2954b84f0cdf8ebfdb5ecb3da7221":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-06-16 15:58:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/991964/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/64/991964/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/64/991964/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/64/991964/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/64/991964/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/64/991964/3"}}},"commit":{"parents":[{"commit":"aa674e6c0a6db0a8dcb82c006d90365d0050a90f","subject":"typing: Add types for RESOURCE_ATTRIBUTE_MAP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/aa674e6c0a6db0a8dcb82c006d90365d0050a90f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-03 13:37:09.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-16 15:50:14.000000000","tz":60},"subject":"typing: Add types for SUB_RESOURCE_ATTRIBUTE_MAP","message":"typing: Add types for SUB_RESOURCE_ATTRIBUTE_MAP\n\nChange-Id: Ib0ea320608cb30e5b094670100ed398adc6e5f74\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b8010afeb2e2954b84f0cdf8ebfdb5ecb3da7221"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b8010afeb2e2954b84f0cdf8ebfdb5ecb3da7221"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
